)]}'
{"id":"openstack%2Frequirements~772860","triplet_id":"openstack%2Frequirements~master~I26083a12cd1bc8dadfc518128d7b5bdb8246c508","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I26083a12cd1bc8dadfc518128d7b5bdb8246c508","subject":"update constraint for tripleo-common to new release 14.0.0","status":"MERGED","created":"2021-01-28 11:10:46.000000000","updated":"2021-01-29 03:59:33.000000000","submitted":"2021-01-29 03:57:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772860-new-release","meta_rev_id":"51c8527c69058ace32df83ca541c500992b43cfb","_number":772860,"virtual_id_number":772860,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-29 03:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-29 02:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 12:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-29 02:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 11:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-28 12:45:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 17:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 23:02:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-01-29 02:18:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3e6a126d90deb85097813ebaf6bb2ef5741e513f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 11:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23fc25c69c5a29c1f196f3279ccd67693442cc3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:54:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e38b8aedfafe42b89d1c73cbc5774657 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":1},{"id":"31a312ca8fc81fb969fbd1162d6b3b005304ed33","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 12:45:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"658de28c93950f1af0699f58cb31094a2d4aadf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:12:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daad64653dbe4cb2940d5558620647d2 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b184092db8f440b82bf519efc609dfa : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ea2922c51b4ee09a8d6e1083fcd969 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb72cc51bb146e18995f40f16b92933 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c1109352504494c83071a7e2262c525 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3af19cdc59aa4020acf6a907cf79241e : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2f9176254d44f6b885fc60c567e37d6 : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7fcfb4c4bef4931b04350beef7d7cd3 : SUCCESS in 14m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95e459db5b1b405d823c0263b9b1333e : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/089126284cc141a49c9d15f29d7e3526 : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a07ef2acadb14d0884601ad73f960b57 : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f220141bd3142e9a51b9e30938e9c0a : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8324fa63c8f4ff49c86225ee71c558b : SUCCESS in 7m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/236ba9715ec3421198e5758fb4337dfc : SUCCESS in 7m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3717c4574773492fa3fac8ec1816aefe : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/109339eca1fa4150a20caa81a4309f2c : SUCCESS in 27m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac39231a7c174a51b635447842efc4b1 : SUCCESS in 11m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/205de955a7be486a84b89dff7143e86e : SUCCESS in 9m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7207d5e37ba749bf94e4509c20146cf6 : SUCCESS in 20m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9f2bcd3a06944eb2ad51e057367d1a54 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9e66feb94ff4bd1a9c0d00460122453 : SUCCESS in 7m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b02c619622234426b40bf32d64c0dc75 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a3c5cb6941f4c8ebd4fe7ef78a8bdfc : SUCCESS in 57m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe77ec877a474d2fa31f580ecd92ab1f : TIMED_OUT in 2h 07m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8e01fb5b3e06f6c120a4d64025108a0e336a0119","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 23:02:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46a128fd3b770c9298235a8a46cb409af8e11077","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8d0d28c255a84636ab50f04fbde00ac3 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":1},{"id":"28a55a0572abb7c0a4777c9cf9ea5413aa3c4369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:52:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d5537aafea4d0ab4f68feaf244e76a : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b2fb37f6af40b1967adfe62a45a9be : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58c92293a2404e8b81d3cbea45e5990e : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f9962f644f6433eb02cfe1721874e4d : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b14059ef0c13426fbf61a9175e52dc3f : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3febc7b3ef534976b1a2f728af74f5f6 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a98c8a060f614a9ca15c25a94d223998 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a7177ad1c70847e6a3c5e69cd6173cf6 : SUCCESS in 10m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6936da70bf4847ab8e2c762c6f435b12 : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2efad431610945f2baa4d88fd62374e2 : SUCCESS in 8m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/caff95e36a63497ab9573a55389d1fb6 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58213ad4fb8044e78803952c508c1f0a : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3642500f5cf544aa82f3604b2291dce7 : SUCCESS in 6m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/164065d658d24de08a5b91b38da5cb0d : SUCCESS in 6m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c100a9bd4fa24498bc1b287e8e6d9008 : SUCCESS in 17m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/780ae29a71ec4a7a91b899be9c803b37 : SUCCESS in 29m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2118d150d60d45b4919a9798d3fe66f4 : SUCCESS in 14m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5aca2f7bf72546f9872c90f09bb9e4e6 : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14309cf08f2e441ea41c2103f49717b6 : SUCCESS in 19m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4a4c091716374e3aba48af94854493e0 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a80b152512944749a2e44d309b07fc0 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9362d562da974400a1eb76e2aa00ee2d : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/295a973d923e4c8fba4fddea450b89a9 : SUCCESS in 58m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bdcf75d2678f470ba1918394702af255 : SUCCESS in 1h 39m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c9b52ef10549b922d715f27c782b35c5958906e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-29 02:18:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa92c4ca5e72a5fb6e7cda8cc5e16752812eaf0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 02:19:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6209f88e18109db542df93b8f524b614b5a06c43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 03:57:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1649745c17e4f54b601ac7d3cbbcee8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8e2c4363ac4984baa5ef018129a609 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96ad2714eb3d43d89371ef056373b236 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c053c268a6b74044a33207377e8325f9 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40de790f7fe9449d9bd1e76c1850588d : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0044cab7314444beb3ac7d5d2fb9df5e : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/68e5c83f0aeb4c849a78233ba810aebb : SUCCESS in 8m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cd8abfa1e693469dbd0b3aa893439005 : SUCCESS in 10m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e37b44c0e09b400ea32b980bb750f441 : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a9da436722d843d6b18c384491ee2c37 : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/52c5c606056f445cb904d592d8228967 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7fcc89ff1f54b52845f6ec9bda498e5 : SUCCESS in 6m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5d546adcca3c4b94a2ff681bd9ad52d5 : SUCCESS in 8m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bbf304c4466b47efb73dc7f6bec9cff4 : SUCCESS in 13m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/365c32e01bd640b095f1e66fddccd2f3 : SUCCESS in 27m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1349aa417ccc424bb23c07452eaa06c3 : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea04e75a476a4fd6b8a60c6914bce8ac : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41e9dd970b684bfdbf603d316a869f92 : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7b67cd8aa60647998c34edf8447833bd : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5108036a6fb04e7f9fcca7aab743ffbc : SUCCESS in 5m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ebe72bd3343465fbf5ec7b9f8b38973 : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd59055f16a342989d87ae2c8e0a7346 : SUCCESS in 51m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c1eab2605744d9b97f569d64174ef60 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f0ec5d3aa0b7c4fc21addcb873b27a4c09e7e7a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 03:57:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51c8527c69058ace32df83ca541c500992b43cfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 03:59:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea59537349a49c89a2c1da4b50687e1 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80498c84d0a270634f41673134213de3fa2327df","revisions":{"80498c84d0a270634f41673134213de3fa2327df":{"kind":"REWORK","_number":1,"created":"2021-01-28 11:10:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/772860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/772860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/772860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/772860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/772860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/772860/1"}}},"commit":{"parents":[{"commit":"cf02df5b2533eb0c69ca3e52d48a856d3b7ed0e7","subject":"Merge \"update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf02df5b2533eb0c69ca3e52d48a856d3b7ed0e7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:10:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:10:42.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 14.0.0","message":"update constraint for tripleo-common to new release 14.0.0\n\nmeta: version: 14.0.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I42b1e3cfa9e22e1a6c483c6c4c96a97a8957872b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I26083a12cd1bc8dadfc518128d7b5bdb8246c508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80498c84d0a270634f41673134213de3fa2327df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80498c84d0a270634f41673134213de3fa2327df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
