)]}'
{"id":"openstack%2Frequirements~772862","triplet_id":"openstack%2Frequirements~master~I959d27ffe1177dcbfb59dc9b1d72c2f6ab4d4177","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I959d27ffe1177dcbfb59dc9b1d72c2f6ab4d4177","subject":"update constraint for os-collect-config to new release 13.0.0","status":"MERGED","created":"2021-01-28 11:16:09.000000000","updated":"2021-01-28 18:33:16.000000000","submitted":"2021-01-28 18:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772862-new-release","meta_rev_id":"1e27cad8f9c2a97300b691007f617172c42c09e6","_number":772862,"virtual_id_number":772862,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 18:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 16:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 12:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-28 16:40:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 12:01:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-28 12:46:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 16:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 16:40:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a30dd7250e755e750bdbd292900ed14f41b2bd22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 11:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2d1f5cb4c33c938276599745f97ff198bb66ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:01:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2213f92e20414d5f96e5060b283d6ca4 : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":1},{"id":"87fb27f2a9198277cee109083ec1eac8e1275389","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 12:46:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3afbb69742a1e0ae9fa0ab199b61a00410a33df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:36:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a656d7b21e437cbbdbfa9b5a30a0f6 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a8ce81b7ba484d93e1cd11ed44860e : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/849ff03ccd6141efbeed90546fe89dca : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbe95f0d868a4d37847346a7a7b7ab8a : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b927ae326124ff3b7fba821118636e9 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a69f8ad0652413ebcc1828d5f7d8051 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/365b144936594aecbece6442f7ade143 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d4fa310866544dd7a1fd71d36864302d : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/030e1b1012e542af9b0bcc8979e2e808 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90034b491ea24b69a0396b8bcdbdaf78 : SUCCESS in 11m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b16b80eb6ffd4498bf726cd20a2ab46d : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22dbd59655044c9fbd55502a9f45441b : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbc7feb04bcf42b18dfbff24eeca0c1e : SUCCESS in 10m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23ede2ca1fce4baea945d253fcf8db26 : SUCCESS in 7m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/077987d9f1fa492c85645f805ea05753 : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3668563350e643abba9288b6067f275d : SUCCESS in 26m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b7f4fe1ee3e4f58aacdeaa4c40158ea : SUCCESS in 11m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5cd97db0abf94f7d8203a6d6a452cbd6 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2e48d645eb84ed5b8fe670006207d8f : SUCCESS in 18m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/591e2c6fbc9d4d7ca2474eeb5c2499b3 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6c0ddab365846d483adf822c66009ed : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e13cebc0f3841838c9c9f0de1c66e2f : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a3271ec203c45b6b5b87d07fb4679ee : SUCCESS in 45m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e28be6395c04131a9b76cc34b67fca0 : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"128ac0e1853736b7860eb9eca6d458cd3cc66cc8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 16:40:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a232e8756ea11a45c9abf8f79a67436565bfae32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:42:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"022ee0dd8f2d1bfb0b56c2b816011e2ba9ddce50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:31:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a604c97741d54fde98b5da6a02110777 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131a649b1520480c81e1721f9dbb8062 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b17988433414bcd9b886fa8d8d02ccf : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f145e680f82d4ddbb8bdac5861202e86 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23739a143c5c437d9380ec2dafbc5a69 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/722d381155b64edc882bd96cdb29db82 : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d11332df8fd4510996238881bcb3e9e : SUCCESS in 11m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70952b5a8c994da5aa146a7b792274b1 : SUCCESS in 14m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2707de72b78d403e9e8ae8d6bb22694a : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2dc2ad3a04ad42d8bf38d67689fb5f82 : SUCCESS in 6m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5bd4716ceda2499e96c8062ce445b2fd : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d221631e63c4f2da800524ceddfcab0 : SUCCESS in 9m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aaae841429264f6cadd6d6087fe626bf : SUCCESS in 8m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/27c477e2196444ba94a9a315085aa9b4 : SUCCESS in 14m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/692897e96589414e9524ceaa372499ee : SUCCESS in 27m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0bdc1dba31843c3a3d6ee2509144b7b : SUCCESS in 14m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/361e90b1f8414225acbce8314caebb37 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c01fc5b3e7584b319102439e519ffa5c : SUCCESS in 19m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/86c3b9c95f524dfebc3727719fe17985 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09eaaf8b2a464473868a52a04f6b1df5 : SUCCESS in 6m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e871afd1ffc846308a4f206c7abab35b : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fe9bed00ce744fcb2f3822c555b01ad : SUCCESS in 1h 17m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60f8e59177c5421cb514ec78fe240e4c : SUCCESS in 1h 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"39dff5f08bd005004600ee982a6b42ec32503a73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:31:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1e27cad8f9c2a97300b691007f617172c42c09e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:33:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3844cdfe0844e4973eccebd1199b17 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c335a346537344d89bc63a4e3d780f22e0f2340","revisions":{"5c335a346537344d89bc63a4e3d780f22e0f2340":{"kind":"REWORK","_number":1,"created":"2021-01-28 11:16:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/772862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/772862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/772862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/772862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/772862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/772862/1"}}},"commit":{"parents":[{"commit":"cf02df5b2533eb0c69ca3e52d48a856d3b7ed0e7","subject":"Merge \"update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf02df5b2533eb0c69ca3e52d48a856d3b7ed0e7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:16:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:16:07.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 13.0.0","message":"update constraint for os-collect-config to new release 13.0.0\n\nmeta: version: 13.0.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I42b1e3cfa9e22e1a6c483c6c4c96a97a8957872b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I959d27ffe1177dcbfb59dc9b1d72c2f6ab4d4177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c335a346537344d89bc63a4e3d780f22e0f2340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c335a346537344d89bc63a4e3d780f22e0f2340"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
