)]}'
{"id":"openstack%2Frequirements~772863","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Id6098591af71656e3f4e3a02733962ea7d2d95bb","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Id6098591af71656e3f4e3a02733962ea7d2d95bb","subject":"update constraint for neutron to new release 17.1.0","status":"MERGED","created":"2021-01-28 11:20:50.000000000","updated":"2021-01-28 20:03:58.000000000","submitted":"2021-01-28 20:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772863-new-release","meta_rev_id":"4f2da19ab682939878a8cacbf009211beec0d65b","_number":772863,"virtual_id_number":772863,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 20:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 17:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 12:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-28 17:02:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 11:39:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-28 12:46:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 17:01:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 17:02:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4be86a17959d5985e82bd2e8300b34808b337013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 11:20:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eea20ead2188afab8a3e66ffac64e06f8d37c6cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:39:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/07ac568bb0b6478db62a3f653c2cd477 : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"82ddcd021266496d6834e754fc836e1860a74c8f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 12:46:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28cae6901bf73497a89630abb5979a8935d73dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:01:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd726c0eb5944279581132ebd8d71a6 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b02a0a4dfc944854a76d25199fe87721 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62c222a198944620bbea17b54fa6363f : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c249711ee7034712a0bb34005be4588c : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b96486e980994c6d805d3c4f86b33734 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/559d41f401e240208557b5f7ee22cec2 : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2549c4468c4b470795d1eca6761b2314 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b596f919e734065bf1413abc5dbb7a5 : SUCCESS in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a761d5689804a75b6109fa61256e702 : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/94c7d7f72d714d42ba5f2514a02c60cc : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bcbee96e2a3d45d88304cd01cddbfc1a : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ade9fb1f67da40c6b5705476e8a9d659 : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6d0f0faff8a4d099635423426a47d31 : SUCCESS in 13m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7303c39bc7242b48855bd01b6dbc696 : SUCCESS in 28m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aebeca5a182148b3a3f74980b30ec33a : SUCCESS in 13m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc299c32d000453fbfdcf5fa20971a49 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14a0e44da1da4090b655dbd11aee23cf : SUCCESS in 17m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dfff71b1db914393b015af6b159c0685 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b58b59d2cb447ff8e17c045010be651 : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/42c9a6e90074488e9c24a06c96b1373d : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dab403cf1464f92bf0656573e55513d : SUCCESS in 1h 18m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fe3ef0bd6f24babba9be7cd426a4b45 : SUCCESS in 1h 42m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5fc84c98349f9cfd2c158afd927afadf78757e9f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 17:02:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b91e6e1239e78ae6763b3fc77d4c50773bca8c0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:03:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9373db081b949653d726766c61e530405c2f5b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:03:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9067e38cd0b490e8dc71ac47e60e574 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2e42f1375d14a6c84142670caa93b31 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/373ad9c785844536b2b47de1dd35df12 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb31a763a7574dc584b996cc03d96650 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98c5e847cbc544fc960285bbcabd2f9c : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a73c00d5aab2475c8d26c6cab515dade : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/295700da1ccf4cc293fe1f77c47e8680 : SUCCESS in 10m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1ad4046443084d48ac58f6283e8463dd : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/121f617f07934f81b9d4e7b26deb696c : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8abba1f8a7b24bacbaa82a56adf1bdfa : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c55e118c303b4f53858a220e9ba175ba : SUCCESS in 6m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a50cc9314ee496f813c5184fa91ab5c : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/885592a289c44c80bc53d8ed5f3b50ac : SUCCESS in 17m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8d79ab9ed11d42fe89f6ec3ff3cc967d : SUCCESS in 18m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/422ba19c618b4213b946097490daa5a2 : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc1fad6bd147485e948825bcc048035d : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fea6e9da9d247c89498cd78f139e6b2 : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c12e03be458e40aeb80828cd670f962e : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f5331b655524736b10bb1f3bde7bdc4 : SUCCESS in 5m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f32b9871290436b81faf4642e5920da : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41dd09fe3a2e44b58d520433662980c8 : SUCCESS in 58m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98a16138c6524da7ab2f7443d1827453 : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4f2da19ab682939878a8cacbf009211beec0d65b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:03:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c96fe0ffc41376c01c51bc1081683db2c65b7c49","revisions":{"c96fe0ffc41376c01c51bc1081683db2c65b7c49":{"kind":"REWORK","_number":1,"created":"2021-01-28 11:20:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/772863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/772863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/772863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/772863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/772863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/772863/1"}}},"commit":{"parents":[{"commit":"5e6ea3feb4aaaf35f9d8365a0a377dd98debaa0b","subject":"Merge \"update constraint for os-collect-config to new release 12.0.0\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e6ea3feb4aaaf35f9d8365a0a377dd98debaa0b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:20:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:20:48.000000000","tz":0},"subject":"update constraint for neutron to new release 17.1.0","message":"update constraint for neutron to new release 17.1.0\n\nmeta: version: 17.1.0\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Commit: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Change-Id: I10f4e1e24486d59217a6a21d11f003c1ace73c37\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id6098591af71656e3f4e3a02733962ea7d2d95bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c96fe0ffc41376c01c51bc1081683db2c65b7c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c96fe0ffc41376c01c51bc1081683db2c65b7c49"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
