)]}'
{"id":"openstack%2Frequirements~772909","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ic1017de2717818161b68a3e9a0f607710a4e8b5b","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Ic1017de2717818161b68a3e9a0f607710a4e8b5b","subject":"update constraint for paunch to new release 7.0.3","status":"MERGED","created":"2021-01-28 16:25:13.000000000","updated":"2021-01-29 01:20:06.000000000","submitted":"2021-01-29 01:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772909-new-release","meta_rev_id":"b87450b506741ae463997e5aab2e779267859740","_number":772909,"virtual_id_number":772909,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-29 01:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 23:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 16:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-28 23:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 16:28:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 22:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 23:02:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"73a6e13af8b72c570886d213ae82f513dd65938d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 16:25:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be2a102075a90b97bc1015de97e18fe44162f9b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 16:28:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afcf2c3f762d20a83d6737af893d18bfc1472fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 22:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b0c8caa04f409bbb78d4b2c589af85 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03160725c45d427bb27c3e98ab32dc55 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60056a3139bd4a2483c149e1cdffc25c : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea23e980cd934d839b12930eb34b2d8a : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e004a77fcdf04b56a856ba9fce0df9a0 : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/05dd684b474c4251a0e8e3d13c5f981f : SUCCESS in 10m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/898a664f96b64c5aa50b667aca485569 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4499fb885504a868882a5f0d586ed3c : SUCCESS in 4m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/01780bdadfc0437e89b7bfe6aab616cc : SUCCESS in 14m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/704cbcb22f904889b214f2dd97bbcaa4 : SUCCESS in 17m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e908653a79747898d9d5b48f61011b4 : SUCCESS in 17m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/396fe63f78d34773ae844d910b6a7c9f : SUCCESS in 31m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ccaea3a86f6648419ca7ec61951815f5 : SUCCESS in 20m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/66a692c8edfa476cb4d163b42799e5e1 : SUCCESS in 13m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24a97c0b9da847509bd0f9cc154f3152 : SUCCESS in 17m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/532ecb45cd5c4e0fba27d847b4dbdcf8 : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ed29b3b7d5c7481ca76052e70b6b3c1c : SUCCESS in 11m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10da7b90a2a54488b4c3da77d9200f28 : SUCCESS in 11m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0830650b1e044e7094938ab4e0d1a723 : SUCCESS in 8m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/debd6649f7dc4aceac2170da37d04e42 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87e23fcf2039423dba7d71cc8e23a54f : SUCCESS in 58m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d934e159f9914ea29f47b8f5c4a7cd77 : SUCCESS in 1h 13m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/330b02b591f748909c84555425ca4a5b : SUCCESS in 1h 09m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/e1d7d12822ac4aae8db05d312c85d3fb : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1d9d42593c887c867e4a7ca60617a9ffe42c5dce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 23:02:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a2394a151c67820fff5bc9f66c94e8bc3024ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:03:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"009d5c7f985501fabf3c5844e9d8d2a65a4ce9b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:20:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f095967cca4f0394a44fcf44043dcc : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30f41c61b4634f4bbf38eac6ad77e5ef : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ae408266f594e7482639527166880ca : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fab00cb76d734dcd8f2cf309d8e38ef5 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c49e6cee37e64d459d9a687944bb1803 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0394fb03b6f34d2e9cbdd7c6e1a441af : SUCCESS in 11m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0985eb26be3b4566bec6e0d6a771e48d : SUCCESS in 10m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc5df57b27184a2ca9e2fdedd02e5693 : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89ab5f76b16f45c8b0b918b4b697759e : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2529256861dd4c57be2f4188710397be : SUCCESS in 6m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/76281ecb739f4eee9203dea499fe3ddc : SUCCESS in 12m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/01fa2712dcb44926af747700b927a21d : SUCCESS in 30m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6f22877e82a94584904bbde3f347828a : SUCCESS in 11m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/759d35e41bc040fc8049ce12689c0678 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf638796b3084e6d9c09b7d898e5a847 : SUCCESS in 19m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/521d2e36314a4d20861a3cfb1dc7d64b : SUCCESS in 5m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8e7b65f98cc74fb5aa8ae5d91aa58783 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d608f66d15a241db857303dc9755bd70 : SUCCESS in 7m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/56d7ee0a5d1a42f3acae652a0c31d325 : SUCCESS in 12m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aab8dc07e6e84ba2af03ab4f4040e8ed : SUCCESS in 8m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19b8d1fd53df48c4b224535d9c285756 : SUCCESS in 52m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87d8088bf2b6448c91b39a593bb9c0ae : SUCCESS in 1h 24m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/604cfa97dfd14abfb834e6f5c3e3853c : SUCCESS in 1h 33m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/3a15b53cc30441ab90ed04927e49d08b : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b87450b506741ae463997e5aab2e779267859740","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:20:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1077be44a8512091f6426f19117bb62c03c1204","revisions":{"e1077be44a8512091f6426f19117bb62c03c1204":{"kind":"REWORK","_number":1,"created":"2021-01-28 16:25:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/772909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/772909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/772909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/772909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/772909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/772909/1"}}},"commit":{"parents":[{"commit":"ee9faac15785d70b264b363aa8e473d6d7362f60","subject":"Merge \"Add neutron to stable upper-constraints\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9faac15785d70b264b363aa8e473d6d7362f60"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 16:25:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 16:25:07.000000000","tz":0},"subject":"update constraint for paunch to new release 7.0.3","message":"update constraint for paunch to new release 7.0.3\n\nmeta: version: 7.0.3\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Id460b7656e9c4e9e00ed19f3bcf35411f668a74e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ic1017de2717818161b68a3e9a0f607710a4e8b5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1077be44a8512091f6426f19117bb62c03c1204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1077be44a8512091f6426f19117bb62c03c1204"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
