)]}'
{"id":"openstack%2Frequirements~772911","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ica96207e986e573a1590236c63b17681f1cdb3e7","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Ica96207e986e573a1590236c63b17681f1cdb3e7","subject":"update constraint for os-apply-config to new release 11.2.2","status":"MERGED","created":"2021-01-28 16:26:02.000000000","updated":"2021-01-29 01:19:47.000000000","submitted":"2021-01-29 01:19:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772911-new-release","meta_rev_id":"45461065a363872481b179503f54795c69fc5dbf","_number":772911,"virtual_id_number":772911,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-29 01:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 23:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 16:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-28 23:02:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 16:28:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 22:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 23:02:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b12fa3a5defe16fe0a8dd54f6c50070b4437e167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 16:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef0280f51660475dff0bdee20134fd7114b9a3fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 16:28:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25fab46f1534c91b821c1b04fa7a93b52e7e365b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 22:26:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ebfde0fd7d4fe18666635327d7cc04 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6f710e1310d4dd8b0585964c6a693e7 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceadcf24eae646298f98187a7c2b453c : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9e811d8c9ff456b937472663c69d661 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df39f267f7554d63a8b8c0a60bfc47cc : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c72f298ad9b14009a5f1adeac952b503 : SUCCESS in 7m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7982907f3194382a64d9b89e4719488 : SUCCESS in 8m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ba7833bfb8c14a1eb50da2752b562396 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a29a06d2e092419bb1cd89d517707be1 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5b6e02e6bb04116b3a89d5973980044 : SUCCESS in 6m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a7da1912c3844c48ac4bd3577b00dc6f : SUCCESS in 11m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4eccff2a501549e4bdea011112259569 : SUCCESS in 30m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3ce0651d0128400f9143be68fd143226 : SUCCESS in 11m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bc745ab79d7949179f2564f92e00db8f : SUCCESS in 12m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ff1d79d6baf4ac8ae26b81c3159fa54 : SUCCESS in 19m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/83480ec42aa448e9ad5450744f2f7e55 : SUCCESS in 5m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0d30a07e435e4ab88a1c9f17830f1be2 : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/854707d20bf541d9a1f43ce7caca197d : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fc357a196148445dbd4c8a4e9325029f : SUCCESS in 10m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ad8d264351054d52890078cc5f4e8d9e : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/487f0ed32b3047a690fc4fa4a47ed575 : SUCCESS in 51m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70f8098a8c344a22b85a6e8f64ed0f25 : SUCCESS in 1h 19m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f8e488a516e4a67b3bc30f763263527 : SUCCESS in 1h 23m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/3346e90967c14725ac5ce630c89ae649 : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"55dd7cb964bcb8343dfdaa8a57644ce901b2c5cd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 23:02:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3505aef40a414a28f649202226d84cfcdb9188ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:02:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62648a29776694d4ac8187a5529c213eb11e5c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:19:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a97189e3aadf4f28b24c217c02c33264 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb551bdc43d84685bff1c1bd5c7813b7 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d211d80b43b4b4fb6b64a9efd35b260 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac13e55ff03d4269b349a2595d22e4c9 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d828af116d214423af06bfd2e758eb28 : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cba1c5e3371f496981a750c6dcfe14ea : SUCCESS in 11m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c7888ee34404e5391b4d4751fe4be55 : SUCCESS in 10m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a15c683c2a8d40108e22a119111be105 : SUCCESS in 5m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90d32dd52a86433babfecc6ebbec527a : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a60c91a09b5b4707bc2e2fd3993e25a1 : SUCCESS in 8m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a11fab2150d44660a313b640f5d63c7f : SUCCESS in 14m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/094545e765774d4dbf9a664c37520b20 : SUCCESS in 16m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e1a01cb7311e42889c7b240cdc43323a : SUCCESS in 15m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fbd256ed324543ff8105f5d1475b73ea : SUCCESS in 17m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/448e4b8717ec467b8f4d9de5bf80fec8 : SUCCESS in 21m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/51f8b8716b3b4c2daee4b1721c56bf80 : SUCCESS in 5m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c7d0924d00ba43a8829151bd3b64d0f1 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15b316b7de454857be883eab365bf3fc : SUCCESS in 6m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/97f9cdd8cdbe4bf6985b78bf09492ff6 : SUCCESS in 17m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85028a67296d419db9dcc3b65b104440 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9adbe2ed96aa4bdb93dbe5aca81fcc35 : SUCCESS in 47m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e13e67d9916a4028824609394cb72cbc : SUCCESS in 1h 17m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f9e95347ba4a3abb1f25c5b0db675f : SUCCESS in 1h 30m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/71fee09f683c4a9bb1061a00f714abd8 : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":1},{"id":"45461065a363872481b179503f54795c69fc5dbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:19:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c94840a169064821a06d8d73887e370111d7bb3d","revisions":{"c94840a169064821a06d8d73887e370111d7bb3d":{"kind":"REWORK","_number":1,"created":"2021-01-28 16:26:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/772911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/772911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/772911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/772911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/772911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/772911/1"}}},"commit":{"parents":[{"commit":"ee9faac15785d70b264b363aa8e473d6d7362f60","subject":"Merge \"Add neutron to stable upper-constraints\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9faac15785d70b264b363aa8e473d6d7362f60"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 16:25:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 16:25:47.000000000","tz":0},"subject":"update constraint for os-apply-config to new release 11.2.2","message":"update constraint for os-apply-config to new release 11.2.2\n\nmeta: version: 11.2.2\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Id460b7656e9c4e9e00ed19f3bcf35411f668a74e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ica96207e986e573a1590236c63b17681f1cdb3e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c94840a169064821a06d8d73887e370111d7bb3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c94840a169064821a06d8d73887e370111d7bb3d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
