)]}'
{"id":"openstack%2Frequirements~772916","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I910477e5c55f12c8bf64d5a0f982a5e1d6c1f090","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I910477e5c55f12c8bf64d5a0f982a5e1d6c1f090","subject":"update constraint for tripleo-common to new release 12.4.3","status":"MERGED","created":"2021-01-28 16:31:29.000000000","updated":"2021-02-02 21:34:47.000000000","submitted":"2021-02-02 21:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772916-new-release","meta_rev_id":"7c362fe32adabe141acf08607cef272465f970cc","_number":772916,"virtual_id_number":772916,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 21:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 23:09:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-02 18:14:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 23:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 23:09:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dce583a64968568bd13d9d5ba6abc58ec2dc8496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 16:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cff149aadf76b5cadccd2cf47162bb0861642f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:03:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975a7b6c261949a39b51d9764687655f : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55e63562f0424ce8a30b7415ab7a7e8d : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/695e68f8d4b3461baf428c24150e64eb : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/508ccc8a83d14710912a9b617c36ec93 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa86ffbb5d904350bf298ae49dfebd87 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ed071cc03cad43678634067315be7216 : SUCCESS in 7m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/605f0fe6e77f4d2c841ea8fbbc5f6241 : SUCCESS in 8m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4fb47cc6a3e84bd3b2e93f1b87b1d2eb : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6fd269fca63b42cfb7130c533e8b1519 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b608d6245cb54009a45a20c58522537b : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f810d23099834f1a89517bff6b0fe372 : SUCCESS in 16m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4c7dfcc091e14da3baed32728cd5d78a : SUCCESS in 16m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5182c4c67914d6b8f323b16042edb23 : SUCCESS in 12m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2853e01254214f2ca49645c7b1a7867f : SUCCESS in 13m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db906bf75ece429491160bb22f428341 : SUCCESS in 17m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1cfb31564e7f44c78ee8488a19b10adf : SUCCESS in 5m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a0eb18bbb62e43df8bb8754840080110 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/996af8e296384030868b64bc851a81eb : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/12db2fbfb0e845f6848486563ee8f902 : SUCCESS in 9m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21ecfffee1884827825e21902cf7fd7a : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcfb040a68e44a3b8a42984e53fd7a0b : SUCCESS in 57m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd476bdf44194a2485fd8c15ac31535f : SUCCESS in 1h 39m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b00f91695e624c7094a852b75256078b : SUCCESS in 1h 21m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/cd3a3e14915b4f9ba460e19372b88b7a : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0ea66096e8ee9afc30c4f1b9a4a60f16f12560c6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 23:09:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82283eafe8bee71d3a2fd418234f45ca2f686c6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 18:14:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f43a43a8a4673ab388cc69a2c8bbb53bd6dd6715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 18:15:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1100b264ebd323e183e31815f7381612fc072660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:34:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d57c90af5cc48e289c9d02ddf4d54e9 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b86170af6ecb48e2a375b39256ada1bd : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/310687a2a59d4fd0b206b39df18e9c6a : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0c6b15c7efd4601abeb752591d5af13 : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/886b787a3678480ba16ee96d972fc726 : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0867fa5b2d84bee9dd2248c5f27b18b : SUCCESS in 12m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59903d8dcb764504ad68c84a3610f4fd : SUCCESS in 13m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d635342e4b14fd4905e598a69bebfee : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/19eafbb70e8a4969b8dde8d26feb4030 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d95bb5efa5a4a77ba0aa0be7398510d : SUCCESS in 9m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3cc21e3e04784dc4996d9a2173b05776 : SUCCESS in 14m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/688aa8b801dd4ddfb44484ce08561655 : SUCCESS in 31m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a74c457ef4b4c63a561905bb8937769 : SUCCESS in 11m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/67fe644b941f4c709a2b567078933131 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f24b68a9c7d745db8e1591746355a5b1 : SUCCESS in 16m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9dfa3840a09f4aac9f156ea1f20f5ece : SUCCESS in 4m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/10403179a78e4a8fa92703517d50174d : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e4778488c0f4af1bd71f874ebe8277b : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d7c954d9e59840c7960caa4366cbf277 : SUCCESS in 9m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24008736679d44eea0d094c207ba508e : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a78291ecd7fb4c2683a9dc84ea7be428 : SUCCESS in 53m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3280967da1c64ed5b88b39f5a6e3194c : SUCCESS in 1h 22m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24096b2971d14553830d12c6cd6f8329 : SUCCESS in 1h 27m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/25caac35c53b49a3a7fcc2a3130beb67 : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7c362fe32adabe141acf08607cef272465f970cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:34:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"997dd1867095ba6a909d90bc8fd90c15e2d063f7","revisions":{"997dd1867095ba6a909d90bc8fd90c15e2d063f7":{"kind":"REWORK","_number":1,"created":"2021-01-28 16:31:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/772916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/772916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/772916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/772916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/772916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/772916/1"}}},"commit":{"parents":[{"commit":"ee9faac15785d70b264b363aa8e473d6d7362f60","subject":"Merge \"Add neutron to stable upper-constraints\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9faac15785d70b264b363aa8e473d6d7362f60"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 16:31:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 16:31:26.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.4.3","message":"update constraint for tripleo-common to new release 12.4.3\n\nmeta: version: 12.4.3\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Id460b7656e9c4e9e00ed19f3bcf35411f668a74e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I910477e5c55f12c8bf64d5a0f982a5e1d6c1f090\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/997dd1867095ba6a909d90bc8fd90c15e2d063f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/997dd1867095ba6a909d90bc8fd90c15e2d063f7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
