)]}'
{"id":"openstack%2Frequirements~773465","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I7412c5d91f7577f4d4ca00a39ae7bf4670d61b33","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I7412c5d91f7577f4d4ca00a39ae7bf4670d61b33","subject":"update constraint for python-openstackclient to new release 4.0.2","status":"MERGED","created":"2021-02-01 18:39:07.000000000","updated":"2021-02-02 21:27:55.000000000","submitted":"2021-02-02 21:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773465-new-release","meta_rev_id":"4cda3b93fdaea4d90da891cab8a05ea4e0e2c470","_number":773465,"virtual_id_number":773465,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 21:27:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-02 18:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-01 19:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-02 18:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 19:22:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-01 22:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 23:31:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-02 18:14:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f25185cfc48e67cb45225750d885112e9d9ed482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-01 18:39:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13f1471a620fd703a3cc8262873d109d1e13fac9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-01 19:22:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b7348bb71dcfc177ef48afb10271cd6e0803b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 22:35:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf3b284ff4e4886b7e665a3cd599733 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bd87f7a150e4560bbc3fd4495ff0699 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3478fc2e86b24089a3b9dc8d759c46dd : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b68e5dd5858043b882a082aa57a35b1e : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a713a9ed84814cd78579ab3e9ced2446 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e43f213ee9d0497dafb02d2388a02e65 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4c8586fbe1614c7ba47a1b072e1a91e7 : SUCCESS in 6m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/75e943cf0b274c57b6d82c7f370ac5c6 : SUCCESS in 11m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b3d7d30ecfa74df1916ca0e04af157f1 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5cea62c64d8a48989412f29c567588b4 : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a867d48bec654b01a83bd0edbf112079 : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5191a65b50dc4b1684e3bef7fe801214 : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3ae3350c1c624bd5992f2f18afaa98f7 : SUCCESS in 7m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/77bd8025a84e4acdbae3fc1926bb1dbb : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/203377e03b0040a1b8abe706cc1f8e6f : SUCCESS in 7m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eac6ecc76fa2493aaec805255521b8bd : SUCCESS in 14m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/565ede4597cf4891bbf4dcb89675154e : SUCCESS in 13m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/86f73aca3dd941438e354653233e1319 : SUCCESS in 23m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8403933523ee4d3ebae82c26f2ecb047 : SUCCESS in 17m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a9847c53fb284fb0b140d57b189b6ab4 : SUCCESS in 15m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/64a2a1e83ca44a328e4ac63112943192 : FAILURE in 27m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/43d2305e5eba40e6a117ddfe2ba373e7 : SUCCESS in 12m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3b6cf434d394df994ec2bc6f2c343b9 : SUCCESS in 21m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/faaaac27f5124072aedd02597fb253a7 : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e430d4b48b8d4558b4d921221c468e95 : SUCCESS in 8m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cef4a54dfdd748d0b19c959ee24a7936 : SUCCESS in 15m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/52699436fd9f4754a81bd600947d9971 : SUCCESS in 1h 07m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dfa3fa9a82f44bb7b3b131eef6cafe7d : SUCCESS in 1h 34m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b05876bc9a87485da297c56dccf88c66 : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9b4b2ecc27e68858f9f9ea0e29083967962f091b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-01 23:31:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53fab8528e8abf382aadbaa7cd66b8e7b1fd9feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:26:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6093cef7eecf4f7988870ce3187728cd : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f01a6da732894518ba78a3e7b8d85c53 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a460698aca0840168cc0a18948c90c19 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66ee9a5e9d45443d925510acf6f4c0ce : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3c9dbceaacf459e81bb3fdedd5037fb : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f3ed16ae59c40ae932664413c4d83b9 : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b07bec1074884901b525ba364574b5fc : SUCCESS in 18m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8bf9235fcf0a4f019d6bfaee442e136b : SUCCESS in 8m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4c85640095fa403da422854ccfc138cc : SUCCESS in 16m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c92f743e6c0c42eab445ee1d15d2f1d8 : SUCCESS in 9m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d4dda82e3011416eaea50229d40eacc5 : SUCCESS in 8m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f28edcc788d74b7fba401e044edc7fba : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b8f5917019ad415bbf5d490eb9b046c0 : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0961c18908d463387a925049a481984 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86337e5e4cbb4f1f961a586bece6b654 : SUCCESS in 7m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c77f85e89c94430ab7c552acbc1324ab : SUCCESS in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adb8a76360da489c8bdbd4e00afd4988 : SUCCESS in 12m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/aa25524b83c14c13a540aaf2ce64f33f : SUCCESS in 33m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f58b3c210734850a568fcba2e5507e0 : SUCCESS in 23m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2b3ba0fce1a147de9dd9a6ef68bac46e : SUCCESS in 15m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c54bbc778a1490b86a523fbef5521c2 : SUCCESS in 23m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b6fc898734ac4728baa5b9bcd77e39f1 : SUCCESS in 26m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fe60f08dd684176819be76853ab6bdd : SUCCESS in 29m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47d6badf06644f0393a601fd41287e69 : SUCCESS in 5m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f0e1d2ce9ca7447180f540fdf9dcc8e1 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d9c5e3b57338426cb62f60fce37f7d88 : SUCCESS in 11m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/547658d32a5d4da79d07f1e015e9b84d : SUCCESS in 1h 17m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/459b36750df0465c8ecf3305a6875f5d : SUCCESS in 1h 34m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dad8ca2f5acd4181b22f7f5ffc36174d : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"edb1077857769b0d4c69603b845e8c7b1349c97e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 18:14:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72f092736a8951c02899f082fb999c0ce09cffd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 18:15:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"341d9123ba068705cd4b2c56e5e7303ea84f7a82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:27:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb220a7f35d4f00b287e99f6f500dce : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aeeb3c3d5bfd47809e4841136cf9ffae : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/877297bf1d34454cb67932563fe64a48 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c806f8f51a524641a1695179ba0cf42d : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/618258e3691d4ccfa93801704c077569 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d722ed5764644b182990e8419051d18 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7358d14a81c2486b912e7ce451f4cc2b : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f161e598e964a0db0a53afea972cf4a : SUCCESS in 9m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bafa5004137143bc827075bc2aae0b32 : SUCCESS in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5a725dba37144cb29c878337a99e1d52 : SUCCESS in 9m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/42e5a3139c03477889af6d6f59f73ba4 : SUCCESS in 5m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a552a6a20a5a4229a4744f1329a476b5 : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0ae3f78dc5f847ba9fae344ae02ffa6d : SUCCESS in 10m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/62bcb2891e184e85a4bb2f575d060c45 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/532d4565545e453ba973d8218ccc4e28 : SUCCESS in 6m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f88cd161b624d0d849ef29b95fda242 : SUCCESS in 13m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b77d12cafbe443aab2ba5fe935430f9 : SUCCESS in 20m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6242e6c0691f43e4b5d198876f636457 : SUCCESS in 22m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c92f7d39f91745ed88ba594f8b42f155 : SUCCESS in 24m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6338d548e98b420a891ae5cd2af0acc6 : SUCCESS in 15m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ca9aea6914049b88289c44e956abfef : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/78aaac34c1504cf59ec537bc9bb8da8f : SUCCESS in 13m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/517bc98c5996485abded318ce00900f2 : SUCCESS in 16m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/683595a6854a4e7ca6a08d41566aebda : SUCCESS in 5m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7acbad68441844059f859b9ddf2f4500 : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bedceadcbe0e49c7a2c30447053a06a3 : SUCCESS in 9m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31d337b6047d40f99272dbef8ce78e86 : SUCCESS in 48m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cfbfed2c6c9649bf837b1284364d4332 : SUCCESS in 1h 35m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/728d916ed4a6455f94ab9dac51ee469c : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4cda3b93fdaea4d90da891cab8a05ea4e0e2c470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:27:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"290f83a21ab652f9bb3b8500e36cde182a7eeec6","revisions":{"290f83a21ab652f9bb3b8500e36cde182a7eeec6":{"kind":"REWORK","_number":1,"created":"2021-02-01 18:39:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/773465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/773465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/773465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/773465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/773465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/773465/1"}}},"commit":{"parents":[{"commit":"38701843b9aef5718239bbb62697e336fe1465eb","subject":"update constraint for neutron to new release 15.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38701843b9aef5718239bbb62697e336fe1465eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-01 18:39:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-01 18:39:05.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 4.0.2","message":"update constraint for python-openstackclient to new release 4.0.2\n\nmeta: version: 4.0.2\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I051b19caecee6af7a287c8052c02966f8fdb178d\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I7412c5d91f7577f4d4ca00a39ae7bf4670d61b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/290f83a21ab652f9bb3b8500e36cde182a7eeec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/290f83a21ab652f9bb3b8500e36cde182a7eeec6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
