)]}'
{"id":"openstack%2Frequirements~773470","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I149cea94d0d7598419e994ea78d6b0db42f5fbfb","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I149cea94d0d7598419e994ea78d6b0db42f5fbfb","subject":"update constraint for ironic-lib to new release 2.21.3","status":"MERGED","created":"2021-02-01 18:50:50.000000000","updated":"2021-02-02 01:08:20.000000000","submitted":"2021-02-02 01:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773470-new-release","meta_rev_id":"a3b5d11f51396a87d7c4f5b87b7e3e7700f8a455","_number":773470,"virtual_id_number":773470,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-02 01:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-01 23:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-01 19:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-01 23:31:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 19:22:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-01 23:00:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 23:31:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8c2279f7d8c9f1bbac881f1f7647000c2fcf7fbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-01 18:50:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7af348e0b4ddd63d57ab86c344576caefb9bcb65","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-01 19:22:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d468a76e4f9adb93151b28ea4c40b1318e114b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 23:00:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275702ea7aa24fafb9e23428341063a4 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cf14cdd1d3047a78ce69408a22ec44b : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b64aaaa75c24513b31a8773ca589714 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c4417000bd3447da2af76edad4e4af7 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a0b602c44d74bf9bd20d18fd005d575 : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b9bfa0e7d35473fbfcfcbb3cf27344f : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/72faa9cfa4a24bc4be6372a155d5d73f : SUCCESS in 8m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b22b2381bd34836bd6e9583edca0256 : SUCCESS in 9m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9bae45fd19a14994b3a28fa94bc7ce99 : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/53afc2af798046ee90244e6f12f0bc6d : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d32148e551d4159a43bd6c829632192 : SUCCESS in 5m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d7640837a60a412a95ab58e8a775c3de : SUCCESS in 4m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6ebfea1a6e5e4cc0965e348983d8e57e : SUCCESS in 10m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/180a81d0e1784ab3a229bd0b533ccb39 : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f544b42bf5d94e5a8552fe976caf1fd0 : SUCCESS in 7m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8beea65e19774c1195f11aaa14305234 : SUCCESS in 13m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dd259d6027264db6afc28429ec028d52 : SUCCESS in 11m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6a80f0c800d14f729ef8a850b04fd40e : SUCCESS in 19m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bbc7c9faf084a938b5f85e604e0c1fe : SUCCESS in 21m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/98330ca3d30f40af988720179b2ec713 : SUCCESS in 12m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3a3e790dfab646099d976f7da30705c7 : SUCCESS in 11m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/014dda3ad8fe47b88f25b35642cee48a : SUCCESS in 12m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3214d053bf8e4894a501a7846dc5367c : SUCCESS in 15m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b03d7b2a34364ef6b9be342cf48ff890 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e94d46f3103644ba94f6334b02c156f3 : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6fdcf43414414e4fb429699bf9842e8a : SUCCESS in 9m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd732503716d45e8bef5e5042b921d2e : SUCCESS in 56m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef609283d94943248b200dd23d770286 : SUCCESS in 1h 25m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5d1ecf7adbcc4a8da9fa0fe4bd269e6a : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e90d67e59c1ab54e31fbf6e6e633affafc06050b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-01 23:31:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa6c7a78bea7854bac429ebb4e7ca42ebb5bbf2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 23:31:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf287ce73140f98927ff4cd4368d8e43f657b2d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:08:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9edd1113a91492b8d75e28c13a49495 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e10509509c2d4f6d9b3809d444d31b1d : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b00ddf67d2aa42979594397020fdfae3 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a52fbadd1335410f98a22e9b87c2b737 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05ba76e59c014423bc324430a6c43c41 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22f61afd775c40a898be96403a68f9ff : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/232708ed22014c64be085a4bc7592579 : SUCCESS in 11m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb8151bbffc642bc911c344ac1541c3a : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ef9e2ff8d6a4411f92be391854244cef : SUCCESS in 8m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4acc3c6c574f4a3e97dcc2ff1f854bbf : SUCCESS in 9m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d8a46754702c42ffbf00accbf66b9624 : SUCCESS in 5m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4115963fb9034ec4be5957ece1ddee6b : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a4c2fda26f424551821a50873e42b7ef : SUCCESS in 10m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/69816fa854f64041ba5b48d2c5074d90 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2a288b5a9464b6894521463a58ff012 : SUCCESS in 10m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cf74460e287e4fcf8d8b4b43a0a96b97 : SUCCESS in 12m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/20498f44c5234cf0a714bb54bd031885 : SUCCESS in 13m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/db1c691fa4e446d292796686119b2223 : SUCCESS in 25m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/55c15673e227450a8f7e4c7a61afaa41 : SUCCESS in 23m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e2505a1bdaf44dd39e949b30d583fab1 : SUCCESS in 13m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe7fc389b71445918bc4850ffcfbf987 : SUCCESS in 17m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b162304888484fadb24251b32fd8e202 : SUCCESS in 17m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65dc53abb65c4020a2556c929b7ececd : SUCCESS in 20m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f25d71c8e50740ef8b27f1f858e6cc62 : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/998245a3f7bd4269a4327f8c6802a350 : SUCCESS in 11m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dec0158ddbd74a138fd1bd6a8467fedf : SUCCESS in 8m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e4487b6f65e424bb32c59a7bf243563 : SUCCESS in 1h 05m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/74479222838b4cceaa3e7ae0ac6fd13a : SUCCESS in 1h 30m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b31b818cfd12473c937078311c0fdb42 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a3b5d11f51396a87d7c4f5b87b7e3e7700f8a455","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:08:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdebd3286f1dca3c00bebcdbcdf3eaa758c23d38","revisions":{"cdebd3286f1dca3c00bebcdbcdf3eaa758c23d38":{"kind":"REWORK","_number":1,"created":"2021-02-01 18:50:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/773470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/773470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/773470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/773470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/773470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/773470/1"}}},"commit":{"parents":[{"commit":"38701843b9aef5718239bbb62697e336fe1465eb","subject":"update constraint for neutron to new release 15.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38701843b9aef5718239bbb62697e336fe1465eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-01 18:50:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-01 18:50:48.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.21.3","message":"update constraint for ironic-lib to new release 2.21.3\n\nmeta: version: 2.21.3\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iefa3f075d513948cdad7895126adfd94a59725c2\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I149cea94d0d7598419e994ea78d6b0db42f5fbfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cdebd3286f1dca3c00bebcdbcdf3eaa758c23d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cdebd3286f1dca3c00bebcdbcdf3eaa758c23d38"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
