)]}'
{"id":"openstack%2Frequirements~773471","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iacd4c30b3a2145a8c048367914d86fd7efa0dd0c","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Iacd4c30b3a2145a8c048367914d86fd7efa0dd0c","subject":"update constraint for openstacksdk to new release 0.36.5","status":"MERGED","created":"2021-02-01 18:58:40.000000000","updated":"2021-02-03 03:24:49.000000000","submitted":"2021-02-03 03:24:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773471-new-release","meta_rev_id":"e6b3113c27bab7202dadd1d4a4dbd67c00cffca9","_number":773471,"virtual_id_number":773471,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 03:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-02 18:14:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-01 19:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-02 18:14:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 19:22:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-01 23:06:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-01 23:31:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-02 18:14:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a9c876b8d1ebcfdd286e7969302baae67bfd272f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-01 18:58:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e0439fc6efa5dd4b037b8490dc6d06df7c5140","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-01 19:22:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae1457ac91ae7ea516acc58e602f3de093d7e40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 23:06:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d62ca94d6f542688ea5f5c367bcd829 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f42d3ab7e524d5d9723a98a25ae8f4c : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6313fb915fdb40fab76fc79359038383 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2cd98ab35f714c07bcaa87f515837bc9 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/807cd2c0e9c144979472167cc07b5701 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecc1e459a9654d6180d1f15f7245a65e : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/193253b5d2e346388c9685c92ca8d4eb : SUCCESS in 7m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3305a82bcc534eb2980ef890f0fe898d : SUCCESS in 8m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bb52c303c161409ebc4d3b60ae4b007e : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e13f52d86ed4468192641fff06259e7e : SUCCESS in 10m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a1a832b75ab9437097f65f5a7fa01e0e : SUCCESS in 4m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1734f50880e84f72ab9deb204168a061 : SUCCESS in 4m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8f87657ded184c188ba2a55250760f69 : SUCCESS in 9m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8fefc4e53613421b8ffd5efb4307eddc : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/093985a3128947dd86ae411020b42157 : SUCCESS in 13m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2d7a100d74934e17bd1fab082a87c065 : SUCCESS in 18m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/973ac2eb89d34dc7900c18e6e3a47b03 : TIMED_OUT in 1h 01m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d4755366b209491daae70a8a4239e39e : SUCCESS in 21m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f4b9ef24a454fadb9bade119e8094e7 : SUCCESS in 15m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/827ea05fde8e483fbd1d121ec7b23dbd : SUCCESS in 13m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7459cf5b114743dd8f2c10ab7c12a06d : SUCCESS in 12m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/42d2dc77954a46ad84b70db99dc257e7 : SUCCESS in 12m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fc8ac3f11404cf5a448f32deaa88ed9 : SUCCESS in 20m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba96473bf90c469bbaf7c2963d014463 : SUCCESS in 5m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/25cb5b81da1947f2ba83ec7e3f5031b8 : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38dd80296e0647acb13eab6fc6da2038 : SUCCESS in 8m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b8c6a6e424c4834855621af84489ab5 : SUCCESS in 58m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bc073baed8ba4f2f9b60ca8a3969401c : SUCCESS in 1h 23m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0a728574b1c342ca96973678ac1c787a : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1ff8cdc01c12854deb31bf286c7076b286b765f3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-01 23:31:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"deac5eaa52ba0663033a77d834018d72eea1b55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17aa3b73266c42228bb043529f567030 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfea20d8832a41089feffd00278ef8b1 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cd7826b91ee4aed8eb6563d7524b2ce : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88895dc396564d8dbeb6f8d59bd6ffd7 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/498e13efc67f4f8c91632b4d08f862d7 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7441224e422e4391a5438e808fec8fa1 : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f86d146e793f4724888e39bb6c300a86 : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28b0ed308b4f440283d8e348f1b4a6fe : SUCCESS in 7m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/55c96934ba75412fa61b576ec4e5c5e7 : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd7b6cf163444c7b92b8dab291ea7400 : SUCCESS in 8m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e7e08df69d764a44bd65a8cac9c1a6a1 : SUCCESS in 6m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d86c2c8d416b473c969cd2787e03ac69 : SUCCESS in 10m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5ee4971a4fa74084ba8312b86c8eff7c : SUCCESS in 11m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/72a75bc0fc44487d82cdf4c3f28a0dba : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6065f3ae88ff42a482185aecd81bf537 : SUCCESS in 7m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/809dacafbd12454a95e0ec335a25c651 : SUCCESS in 14m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/596f0c6e923740d39c0049b13a10828c : SUCCESS in 18m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a49db8a87edd42cca814cad13cfc8d04 : SUCCESS in 24m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5b7290f2a3fe47a08e0b9d7d6dcd68d5 : SUCCESS in 24m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/009fb726c3ed4c39a83dd292f8af050d : SUCCESS in 13m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/58d2beb2cea04b57b2e64c71b911b6bd : SUCCESS in 13m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e20969b8041a4d0d8a210d18181221a1 : SUCCESS in 13m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc9e53277b894503874562856b96fc9b : SUCCESS in 20m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65a57c0f2b014ac88bf41a2a9e815cd1 : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/087c623917ab484589d7d36df9644675 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/45f6102cf16f467eaa3bc91a73678c57 : SUCCESS in 13m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5def5169a1e541e3b7aaccf3bf46839e : SUCCESS in 1h 00m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b0201563aab24dabae1b82373e8e9c49 : SUCCESS in 1h 31m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/950c9824ef5a432c86f8d8afd512bddd : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0b96ef7f1e246a6edd9b5b3d7002b8df73c53b16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 18:14:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"00bca75094a71faa553d7f965a40f9c34b845b5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 18:15:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7a9f4d3815c40ad89613ea2bc3e5d014c37e819","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 21:27:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/491d941794a9412eb9035673b1cb5bda : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f0bc92a9e6e4e18987f4d2dee7b7f3c : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9c89d67f2bf4832af19a7815c454347 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28b3757c11314a6ea21fcc602ac892e2 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a14047a1c52844cc8e6312c4ec9649ef : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/64687ab9fd004b20a789102de8168ca0 : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5b9e3797b5b54f749173356edcd9e7fc : SUCCESS in 8m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7e31c35b93ba4061a2cd85320e23efae : SUCCESS in 9m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/467c852f516b43ed8f144a6ad5ffeda5 : FAILURE in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fdaddd49001f489d9e2a1777bd08fcb8 : SUCCESS in 8m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/84b85a235e1c4378b4983c77577265b3 : SUCCESS in 6m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1ead9edeca694fcdb465ce3139f63e25 : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/45d0273b608f4594bfc7794a3596d70f : SUCCESS in 10m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b035a7af67734b92b6f5702bfc4d8359 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85da146a5d124cbdad2b86f9a3209849 : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/32cd3c3fcd9e48d581dea579a23724df : SUCCESS in 15m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db1a14232bfa434c82a8db171819b22f : SUCCESS in 14m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/36ebfc7286b94f868299aecd8266e47a : SUCCESS in 27m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d34103e962c548a08956e3d37789d071 : SUCCESS in 25m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8d2d6df3612f459b8b75daf3ed536460 : SUCCESS in 12m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe64999a212b43eaabfad9423c957573 : SUCCESS in 12m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b798a4003c0f425f8c275b7382ba29c1 : SUCCESS in 12m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9edb73830c0f4818a3c00b606667c995 : SUCCESS in 16m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45d2cbb2cdb944deabeae29697f6c468 : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d50cddc0354a4bab92ea9756b632b016 : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e485be0ca69a41239d1d687013a95f89 : SUCCESS in 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ed4513730dc4f819a1c0d9f7f64dd9d : SUCCESS in 1h 00m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7bc03c5161d4a4394000bd78237bfd1 : SUCCESS in 1h 25m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e8538b00e6a84c239993985fb03969f7 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":1},{"id":"43580b96547694493c5fba5309f589ac1a250dca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 23:50:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2725f42467cd45cc3e0c14a4b2e589fe8eaea2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 02:11:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6b836743f84ead85948937b804c920 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2b35c5faed44e4d9008aefb56077da5 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cce29e9507744712bd463b19a97eeb03 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa7085e38cd84339a6a308bb7697c98a : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78663e79780e4c708d81aa515e64594f : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fda823e06fb436da1f24430e1e19774 : SUCCESS in 4m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/36e4d2b67ac64d6ca278c7ea5809bc94 : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa2582c491aa47d7b745806cc1cda09f : SUCCESS in 9m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/011dcb1971284161a100502350453653 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/511cb537887f4cb3a6d9815fb575326c : SUCCESS in 8m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2f2b885c67c944f192122737b867c294 : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b1b9a782b3d54a28b7ba36d0a5e24337 : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/52008d127d8942aa910f36f03ddf1582 : SUCCESS in 9m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bdbf3f72e66141909daacb800d5447a8 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17d27d2e8a34400a9ab6458885eeae10 : SUCCESS in 6m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/889f55d39aee4906b947ffd117cf5b3d : SUCCESS in 13m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0111ab9ec9f44d358d6973b8dc78b293 : SUCCESS in 11m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/deaac73e60a04dd3a5b62a6d5f317071 : SUCCESS in 23m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/01c3b00964b649f2b5e15658473af2e1 : SUCCESS in 22m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0f3421bfa0294d43ad0b19d232247be3 : SUCCESS in 12m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d675097349384790bd0afc3c1addfc44 : SUCCESS in 15m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2ceb56daf101482fa63ec9b13693ad8b : SUCCESS in 13m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd013fc5cb1b452c8f503ef75c73a29c : SUCCESS in 19m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39d5db1753a94a079d578cdf3a5c7154 : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c278a72db3884ed9b33bcc08ec54bafe : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/64c4b24c453743189b998270f90945e4 : SUCCESS in 9m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/439c51e0f0654c48bd21941be1378698 : SUCCESS in 1h 05m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a0282c7c3304aa89560f131482738e7 : SUCCESS in 1h 30m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f7c462dccf674d849029f7d221b091f8 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7525e8a7297b3090e86325a026b7282d28cb2766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 02:12:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c072aa23c021a51516e154363674e78681bc6d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 03:24:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ca353fb3c24b48a1a48de20b7cc884 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa50aa800f084ea99cbf12821500530f : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aca007d52cb4c27aa68287ff1b031d2 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6413cd4083464e8d91f18d8bebc1a143 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6b0e328916c479eb88d1d490f8824cd : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/908840cb8bd04fa89d520f63c339f7f2 : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4597687e90ac431d95078dedf530b8bd : SUCCESS in 8m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/753c909a27c44e43bd849077a819e82e : SUCCESS in 9m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f47c200b62d545d589734c212ed50884 : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dccf0c1c2dc44aa3bb300daae0ecf445 : SUCCESS in 8m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/97f61eeed5884eb182efcc889d2bbc36 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/80d99b442534447c9bb23b570e1f3e8c : SUCCESS in 5m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7de80fa5130b4049ab07e7a778d3f889 : SUCCESS in 8m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22f6a667871240488fede3ad74699a6e : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce7f12e8cf2b4c5bad60dbcb5ebae1b1 : SUCCESS in 6m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/381412c37c124bee9de5e0a14198faa4 : SUCCESS in 12m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/29032924c71946ff9e1b88937ee5a263 : SUCCESS in 13m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a3a816ef8b774c60a5050f9ee042d1e5 : SUCCESS in 23m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52d61ed6cc834de589c1abdc198b1d37 : SUCCESS in 23m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/58a806a32bd7423eb548ed68101a5492 : SUCCESS in 11m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9981e0abbbde45f78ca112fdd1f16b49 : SUCCESS in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/27281babb9b84750b22ad581a93cab8d : SUCCESS in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6f53e4168794c68ae2254aa05f80a1f : SUCCESS in 19m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/398245dc06a241adbdf420c378bd2b4d : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4bce6f977e3446049c2db7e3edfa9511 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/303d8fa529304372b97077c835b596d0 : SUCCESS in 8m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a29a1a0ff89642a2a01d032d279ba693 : SUCCESS in 1h 04m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa0716d59f064e09993f92fc099d016f : SUCCESS in 1h 10m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/162d22a5d0f24809866ac717e1212d99 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e6b3113c27bab7202dadd1d4a4dbd67c00cffca9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 03:24:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be9b4bfdaf876338120a313b043bb1fdd0ac7b68","revisions":{"be9b4bfdaf876338120a313b043bb1fdd0ac7b68":{"kind":"REWORK","_number":1,"created":"2021-02-01 18:58:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/773471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/773471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/773471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/773471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/773471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/773471/1"}}},"commit":{"parents":[{"commit":"38701843b9aef5718239bbb62697e336fe1465eb","subject":"update constraint for neutron to new release 15.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38701843b9aef5718239bbb62697e336fe1465eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-01 18:58:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-01 18:58:38.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.36.5","message":"update constraint for openstacksdk to new release 0.36.5\n\nmeta: version: 0.36.5\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1f0b93762181141626c39126cdaf3386819a1735\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iacd4c30b3a2145a8c048367914d86fd7efa0dd0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be9b4bfdaf876338120a313b043bb1fdd0ac7b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be9b4bfdaf876338120a313b043bb1fdd0ac7b68"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
