)]}'
{"id":"openstack%2Frequirements~773573","triplet_id":"openstack%2Frequirements~master~Ifd770cc8f35c7ebf97f831ff9e2f1d92444b5061","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ifd770cc8f35c7ebf97f831ff9e2f1d92444b5061","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-02-02 06:22:33.000000000","updated":"2021-02-06 03:59:17.000000000","submitted":"2021-02-06 03:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773573-openstack/requirements/constraints/noclob","meta_rev_id":"498406f025b4a727c0e9b27c029d65b33fd240e6","_number":773573,"virtual_id_number":773573,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-06 03:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-02 07:19:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-02 07:19:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 06:26:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-02-02 07:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 09:36:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14d7ab66d6fec44da45d3b58939765e290ed91f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-02 06:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d530546f33a9dc845bb379d93a4ab7a874ce9fef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 06:26:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eccd1219501ca69730555c21abcd98bcff0a0b82","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 06:26:59.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"672dcf94d552e78c9560af7020d59057a16836ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 07:11:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ea8f210fd8ab49c08d081426013e0c32 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"71cd168326bc937dc11b08dfc6eb0fc3fa5c3a14","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 07:19:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4aa6f1db5fb0656a20bb68ba10d6045e67417ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 09:36:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b05f058ef1434696c9d92064904390 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259c086fb4dd4f70acf281520a6130d2 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a9dbea286fb4757a011e888e8dac0d9 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8a22b0fc6e748cd9223143433186025 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cc2a4a3027747858aaad3f2585e43d0 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/633ac7ef3dc44a90b6cbed307fbe3147 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed381b4d0ee74a64b6d028bd06fad209 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a2592c9da19042e4947797c477d517fc : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1cabef8306e448a6a0b71dc43fb3ae88 : SUCCESS in 7m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9cc5f60babe64d2da4b7cd0753a8ac3e : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f99ed436808044f1ba4019ae7faa4bb2 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48762a183cac462bbed7d1d846cd5771 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4490dca4f555466aba85b8e4bdd6b87e : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/464c72f5a99b4d30a0bec2237cbba054 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d9ff114c80043baa24e0a58eb9b3bf2 : SUCCESS in 26m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/91d7775b2bf541e4905c99ddb4b14f95 : SUCCESS in 27m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e4b208ea082f4978972bc540785c79d4 : SUCCESS in 14m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dbc704192a74b26a17c24d37d0ec763 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9881fe1b0604421594ce5eb2eacff925 : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1ee6d9e6cf5d495e90ea9ca378f6b6ad : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a1b370900af4208bf2e74a173867bd5 : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2dae5c90b71b48fda7e1e77b5736575e : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4180359f32244ada0288edcbd74d1e2 : SUCCESS in 1h 02m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4333614342d4fba99bd61b46e7ae81d : TIMED_OUT in 2h 08m 41s","accounts_in_message":[],"_revision_number":2},{"id":"341426aa60eaab46d0a516b5c0d507de34441467","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 21:51:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d29b85fa1b6d88f6b36ecb322696d154de28fc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 22:27:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6bf4acdcce93460cb29910f4bb508e19 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":2},{"id":"01186d66acafed60af90ef2ee198513b790cba29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 01:31:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bbfb9d3bf584ce29ce6c133754d73b0 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea09a736f5d344abb460da3bac1e34fc : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f44cb14b314c484a83c576a67317ca90 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3da4983e32904500ad24151feea2f0f8 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e46aef299ae445549bba273aa50b7555 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db6684191eb14710a1b545e9df2fbcd5 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d994e8e06624ecdb917d6c2725521cd : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/adf614dc78144dcba7263642486eb1bb : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30b29f88ce30435489a034978e471e15 : SUCCESS in 8m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e82f1456ce0246afbefe5b33b7369ebc : SUCCESS in 14m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/292d887a76b44a8c970590e4e5b21bcd : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ab656de3bf949c79dc87abe7b144f96 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4c88eaf13b945db9f65dc92ad8bb769 : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/79592288dbc746c19d21509938412341 : SUCCESS in 6m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/59eaac00f9bf49d480f0c34d7d13cd07 : SUCCESS in 12m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/885f93bda9db49c58e899a6058aa6bb9 : SUCCESS in 24m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce5d61c706ba419f8ec694cc3d4c3706 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98ce6cad59e446afbfae357e49a06050 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea5540c8a85442ebb099d2f7452e0893 : SUCCESS in 17m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01e2b49cfe7c473e8324e7f15bba6aa4 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0006a58ac5c2460fb82a2563f99edfb2 : SUCCESS in 6m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c70425fb396a4dab93b6df6370ae074e : FAILURE in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ac02a35e7394b03bda9648ca799c31e : SUCCESS in 1h 08m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2933822999c4d648be6b0ce3f3283d7 : SUCCESS in 1h 34m 23s","accounts_in_message":[],"_revision_number":2},{"id":"687ad5eb67eafd956c9fe01fca37f52c113c1ad9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-04 16:49:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"90f77a2d8199edb1f14f66613a9f7390788d3e7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 17:25:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6ba07c1bfcdd4e78a721af8904d10cad : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c2062ab35efcd53bbdf6e85331d818c1493a6c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 21:43:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a895cc14404a5f9db283360a8cd198 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5312800324774335939b7ec775e23288 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4705a2ef46c6413785fa5cdc47ff61a8 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/598434e9fb4d4783b2469ea3d2aa0c52 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ccf569a018245ddad7116b9575ac4b5 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87a3e05a29c0460aafb8563074b77e83 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c12625bf5bbe4ec2b04592e847040a5e : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b834ece803447d4a1a16b90ef6dcff9 : SUCCESS in 9m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/66b9ab6c2fbd438cad41d8ea6db1af91 : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ce5c6fafe634fbc843d3a5f9d7c3483 : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e77d3d0313f440e1ab3d185a2ea9b8bd : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e528e47bf7da46fba26e4b3cb2c7d704 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22f4f3807472414fb2f4a3936f97c766 : SUCCESS in 6m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7a0c73416b814432ab74f2330d873d74 : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b8cf83db7634e29a60d8919efbe85dc : SUCCESS in 13m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0c093be2c9df42f8bc10dbcb4655c9d9 : SUCCESS in 17m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46471d560680472ba9e9e11aa64ff80e : SUCCESS in 13m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b88795541ca4fba870b00e039a195bf : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a16fa1254154f4bbd80912fa29dfb2f : SUCCESS in 19m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/06dda7f8264f44dcaac9b7b139dc6865 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a680d93063b54323b3f31b98982a7f47 : SUCCESS in 5m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/41490fa493984851914b795020e3bc30 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9070fe06dfff47bfbfe3c832abac2546 : SUCCESS in 1h 01m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8bec11d7383467f910798f144ee9df6 : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":2},{"id":"bd8a30b60d335a78ba648f3d066c66fb19369e29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 21:44:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56e2a78ce90dff1cb68c87a8c666babc53adf5a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 01:58:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/999021cf217f4c0db8d974743fe19190 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0a7f93980e4787bdccdbf85180d212 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed2db8d40ed4fa1b477a87a137238a8 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0512121f07747ed873bcb413a0f4a2c : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cd7492f9ceb4bce96e6c264721074fb : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78982cc17ed2462bb4ed843477115c35 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/42b890f9d54b4058ba90b406cbbd993b : SUCCESS in 9m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fc24149105944c6aee147908ee0ee36 : SUCCESS in 14m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6965b23f5dfd4f2d9e9d5223103c7f37 : SUCCESS in 11m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8e915faa41ce4e08a7dc2bfab91542a2 : SUCCESS in 6m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/100112504a1c43219b7b568b094d75d3 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fde60101e26847609ec711c4a50390bb : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d6f56f35c1e64338a400350ef0d593da : SUCCESS in 6m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6b883b160b894b7f955a62503f6f87f7 : SUCCESS in 14m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f3bdea3492c4fd6bffb85d99fea06db : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e89b4dd30e8944bbab6bff66e0f112fe : SUCCESS in 14m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/153a4516d1234105b441cbf31f7068cc : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/539bb43c77d54efe8d6a81bbb9ef8123 : SUCCESS in 18m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/16b083eb82fb4a109e7de5f599d093bd : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb0ddbdf45e84cbb97d49109d5358510 : SUCCESS in 6m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f6e310c5818048ec94a97b3cde0f0b55 : FAILURE in 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2dda7ce0b594f14bce78e3628fedca0 : SUCCESS in 1h 02m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58794827a3c545ae9e2c1773c2a8cc7b : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f2716e04d4f9abf7f3d450dcae0628fb3288b531","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-05 05:45:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04d4c41ed3a9b95df7acbb0cfd25b95618449fa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 06:37:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0a3050f746f04d7cab4f5686033bd02f : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":2},{"id":"598f3120d03b9972ecad5ff57cc3acc3e0aa6473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 07:01:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f49b058a85b4fceb8d53f3c7de5ea9e : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47316f22a7534ac18dcdb9f91cbe4b93 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35f6d9ffd6ff48c7bd8ce0ce7e883154 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb6b0ad62bf443029d8806bb9ffd4e6d : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6b0817a1b3742149a2e28af67e07323 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e07b5377bb4343febffb7f79f18a19dc : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80279dd5934c48cbad01a50f76937f21 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a2e3a988bed4d27bb8b9e08d899b190 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/286e85b754fa4ae8bf1951e7f33f531d : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e66ef57cb1f44f6a0f3cab4dd6fe05a : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e08a7aab8b73485ca14fe721f4d52aee : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0e03dac047a54d4a91a5213c517a0af7 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d03244535144fba89b15c0303a7b98e : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9301307b931b4e30bfe38d058b5ad344 : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5e658b2270eb4cb4acad98a7075a1260 : SUCCESS in 14m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3e8cefa0638b4cc2894127482ea42490 : SUCCESS in 28m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49336f0b89ed4353a36043de581108ee : SUCCESS in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6b61b3e33f64a4da1790976611894f9 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c92759b8cb92493ea3aa7479dffb9e0b : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0d72a4c347ec41738232cfcf7a9f6b81 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4432f1c3c1114650862da65f41a5820b : SUCCESS in 5m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/735d4dcb82cc465493f73c545c7f1f23 : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b8c12fdd4db4a1c8f9c29468b93431b : SUCCESS in 59m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/800529f148d24e6989de7979b6669715 : FAILURE in 1h 10m 58s","accounts_in_message":[],"_revision_number":2},{"id":"603bd90c02ee2dee0563f78a71bdaa906947dd33","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-05 15:27:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"810c5c4b72c76f4d84f8686efaf0d8260bdec0c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 16:07:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b81e808a94c4ee5a549b300d1886e82 : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bfcf5fd20977677333bded004871db1cf0665328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:27:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5ba441f55942dc960eea0281d7764c : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a568718a8e3451b8b0a1bbcde6acebe : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa09bfb69398404f931283a4a6cdd75a : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c41bfaeb858547c6a54db8a1206a8395 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09e1747cbd6d4a0ea502d26696689876 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/782d381fbc8143b7835b124fbaaed1c6 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3fa1d8ac80914234b51808f4a6ac5ba9 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db6e8461c49147dfadc725dc8ac28186 : SUCCESS in 8m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84de867d2ea34e2cbda14ddf5fe90e41 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d25e6fa20d314cbb8ca3aa8effa190f4 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4f26e6ecfdc453798b3294c3ac0bbbf : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac9fc6b675ab4a4dac20c0a5361951fb : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea9b4a581e894964aaf6bf645532a536 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d1bf727bd11a4454a8fe4f87c6510aa9 : SUCCESS in 18m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c5f16e7e448a4c3eadf18905bf9c7af9 : TIMED_OUT in 1h 01m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/040c3f30714549e2ba8b2cffb16020a4 : SUCCESS in 27m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/852267a2602e4ff1bca2fc876a761cce : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/368661ff68284d3da0e733357d659266 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/833bb8753a9940c8b20a39b11c79d899 : SUCCESS in 18m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/85828cbb8de14c71ab3ec2b36cc2a1e3 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a130a5a83aa64cd49d97224a87a66b32 : SUCCESS in 7m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c5494c9cc2f741d9862621bc6c9ebe60 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/952174ca072d488b854085ce2feb96b2 : SUCCESS in 59m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7683caabf0847aaa9f9ced12394b61d : SUCCESS in 1h 36m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4a3547e741735d28db2665d16d22963a30d39969","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-06 00:58:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7766d445c752d91ba969fc8ecec845bcf858420b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 01:29:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b3cf78f6d1c444dab128f27d98a5824a : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2c80c231e3d625e36ae8a7a3caaeb6e9f1d36cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 02:40:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5a21b8178c456c993e054de60f4a4f : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7889438f36e488d9e189c03cbdd26ac : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef561ebec25847f09539a78093a8a6ad : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618ed2e61e3d429196bc41fca4a953f0 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2944cf30846d47058ff1882449363fa3 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d97db518454e4692af2f36ad35ee28a4 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5cf143da44d54438a1e851767a57dcf1 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5925701349142eeb61aa70fd4c3fd8c : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/791b2e22a9b34fc59edca1cced223ce9 : SUCCESS in 14m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b9d9b2c28419470eb63b7c623b351b99 : SUCCESS in 12m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/921c93fa7960495aab07e9ed2ab423a8 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/558728aac6264d44987f07615f261f76 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f67452b9dfab42cc97d3675c72d8657e : SUCCESS in 7m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15a71ba37f584321b7c6f03ff5c5a0ee : SUCCESS in 6m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ddf6e3c50444954a195a33250827522 : SUCCESS in 14m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1a2fa0e0e3d348ccbc1a2d8bcb858f7f : SUCCESS in 27m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/768a4f9b65b54e33aa48aafb27f6ac3a : SUCCESS in 11m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4285ed6af51464db11b9aca2d53d792 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dd802f14b2c4c5fa484566ffddf88cd : SUCCESS in 16m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5db17029dc8c438e9a4ad54386d3adef : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff47201eb74346059c247900325fede7 : SUCCESS in 6m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b6babc1afa794db48806de6b51e056d6 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb0aff31a3754237a16309212b96f733 : SUCCESS in 1h 01m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01f15395a4da445a9627d36b4e3bd576 : SUCCESS in 1h 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"13c1e563b51f6c1abbcf6633576cc07d70deda42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 02:41:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"909460502bd144e5f846ffaf0e34d9acbde7afda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:58:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc7c5996a9c549b987b2fd73eee71983 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504f7953e0e64864945669d7f1218918 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f1c5a1f56ad45a4bb3a26381f45b5e0 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4db527bbf5da417e85495f2cb697db13 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d26db2a0135c464aa67124e590a28d63 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca1de32c9bc74ea5891cf4c7ad26032e : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71c42d2ad57a4cfd96c3157275845f93 : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/065e6557861e4718a4a3d0453bec3857 : SUCCESS in 11m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0506c86cff7945939a4c5ce1524ee262 : SUCCESS in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2f4e41385884a98aa4aafd562a8b5a8 : SUCCESS in 7m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/78f75145d5c449d1907e7e8a916f32e9 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26d94f4c34924f649bf8a4e0170057f8 : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1407c8e5042f4e2c8a599ea09aa7746f : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6c37bbcc483049b785c743b83ef5bd6f : SUCCESS in 16m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a647ddd9c28745deb13aa3f2c25a0dbe : SUCCESS in 26m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/39ea3924c38d4dc381d568deecb571fa : SUCCESS in 14m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/213308aad6084e1296d1eea2e1c20deb : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eecf768973d845d4a9b746d5347a4968 : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a1e256232574f58b23a4b92fa4ddd77 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72f72e2e9d0648c0ac1834de76d4a403 : SUCCESS in 6m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83a40b8dfe54421cb50e0c76329cbc7c : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7184be3854e44bd288ac9f051cf21c5d : SUCCESS in 49m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e47c92c5ea534f269e8044806ce87246 : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":2},{"id":"63777d3f5727136c7c7bef964d73c72788945ee3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:58:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"498406f025b4a727c0e9b27c029d65b33fd240e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:59:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0bb455ac1b49a2862686fb0f9e9f1f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a05820984972d2540ae229f8771c33a67e72336","revisions":{"d57c723d4768b36693ffd6836406e73c90552a79":{"kind":"REWORK","_number":1,"created":"2021-02-02 06:22:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/773573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/773573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/773573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/773573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/773573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/773573/1"}}},"commit":{"parents":[{"commit":"0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-02 06:22:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-02 06:22:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifd770cc8f35c7ebf97f831ff9e2f1d92444b5061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d57c723d4768b36693ffd6836406e73c90552a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d57c723d4768b36693ffd6836406e73c90552a79"}]},"branch":"refs/heads/master"},"6a05820984972d2540ae229f8771c33a67e72336":{"kind":"REWORK","_number":2,"created":"2021-02-02 06:26:52.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/773573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/773573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/773573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/773573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/773573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/773573/2"}}},"commit":{"parents":[{"commit":"0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-02 06:22:30.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-02 06:26:43.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ifd770cc8f35c7ebf97f831ff9e2f1d92444b5061\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a05820984972d2540ae229f8771c33a67e72336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a05820984972d2540ae229f8771c33a67e72336"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
