)]}'
{"id":"openstack%2Frequirements~773695","triplet_id":"openstack%2Frequirements~master~I78db12410381908335b86d27d34ec8e7fe594853","project":"openstack/requirements","branch":"master","topic":"cross-masakari","hashtags":[],"change_id":"I78db12410381908335b86d27d34ec8e7fe594853","subject":"Cross-check with Masakari","status":"MERGED","created":"2021-02-02 14:38:16.000000000","updated":"2021-03-02 07:05:44.000000000","submitted":"2021-02-26 23:35:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773695-cross-masakari","meta_rev_id":"aa95fabb4ebadf9140a219b0f84c8a4c67ebfdc6","_number":773695,"virtual_id_number":773695,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 23:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-26 15:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-26 15:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-02-26 15:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 20:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 06:24:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-26 06:34:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-26 15:18:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-02 07:05:44.000000000","updated_by":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"reviewer":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"state":"REVIEWER"}],"messages":[{"id":"49f872c12542cd962273c3764c2344ada2e284bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-02 14:38:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a573adcfbe36d40abca3d38e7c5ac35b7adca32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 20:46:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75d8352ce104a4d9530abf9778bbd39 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/999e51bd42aa48f1ba88aa43bfd8a47b : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb4b4138bca4970b729ac8b56aeaaf4 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eec44d6baa56447db58ddd7e1ff7a930 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0807ddbbe880480d9fc766081edf444c : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4e4234be0784e20b2866c80c0ce562a : SUCCESS in 3m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52ce4b6458c04f6cb03e9e5a7bb3165f : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bbe6d4ad56f34809b637c8ea8150c6f1 : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0d3a5df7d0b346de96588640126c9956 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/630dfbca8f5a4c8d84642cafb87f8777 : SUCCESS in 6m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d459bfd02ee54c5da9bb2553ae9e104b : SUCCESS in 12m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/583852879bd54f5e9428139b95bc5bd6 : SUCCESS in 14m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7a832631349046d7ab6bd8f4dc29cf63 : FAILURE in 4m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/46c857e024ca4f2ba3bbfaad7a8f1566 : SUCCESS in 31m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a83ce2677aa94f22ac08eb7941fdaca4 : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd8e54ed95c346e8b5daede17787613a : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8bf11e0af7f45af8e475faaf54ab55e : SUCCESS in 21m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9e0014e8810341fd89812b909ca78438 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d2de9e8339c4e9ea65409915e4ea6f1 : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69601d86486a47beb2f9679d8d2484f1 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/507b09b9242449bba4450795c1298c3e : SUCCESS in 48m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2478e5d1df2e4704b3b376d5bc71668a : SUCCESS in 1h 50m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ce1a9b62b0b809d20ab9e9917f3d4f5a307d4330","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-03 08:07:51.000000000","message":"Patch Set 1:\n\nyup, that would help","accounts_in_message":[],"_revision_number":1},{"id":"5c67a8b0cf0b02f853a0edce9f0208b4c92fb8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-03 08:07:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"217bf20060a80bd4c9e06bb44c5249af8ea0e1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 09:40:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0878c9242e234622a335e9788dd18890 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7172f0d6be58438c982f8794a4e98012 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf9be704b1746f9b2e83f293cddca93 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa5a7e2e06fc42dba5785777d5c15dff : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd8c24bad70a4aba835f2c20744610f5 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca8e1dd479c248beae4017be539b537d : SUCCESS in 3m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ba12a25a1ed48db8adcfba75432fab8 : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ee02eb033e74ebe9d90cefdaddfe2d8 : SUCCESS in 4m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2217ac9988374881a16676480df5724b : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55b96c9ec8584021903afe04537f693f : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/03f75cef936c43c3b24b7bb56858a221 : SUCCESS in 7m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4bfa30bdcd194ee394d7dd2b42ff4c2e : SUCCESS in 16m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7934a46ead69406d97b60c70053aa2bc : SUCCESS in 5m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/715ce4c2895b4c4a869b1146ab4c4a1f : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6480a223f9a94760bc63f378291db38f : SUCCESS in 14m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d7334f5d0ec4570acb5544865999efa : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd63bb97c57646cca48266e9fefa792b : SUCCESS in 15m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c47fb7c7f4604e05b7653cc4a44b625f : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb93e5020a9346198df2f79e87e1aca3 : SUCCESS in 5m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1c6f3797805d469ea1bb668b43779167 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b54a2cd1c4084456badc31329301d4ed : SUCCESS in 1h 22m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54bb9d9963f84b84acae45514097f7c9 : FAILURE in 1h 28m 38s","accounts_in_message":[],"_revision_number":2},{"id":"391dce86e602040b374670d7309b9483fc9f98f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-05 06:24:01.000000000","message":"Patch Set 2:\n\nhow would this help (libs that are covered by this new cross check, citing previous instances of breaking tests downstream could help as well)?\nI like more testing but want to make sure it\u0027s doing something we are not already doing.\n\nhow often does this testing go down?  Reqs is kinda the central gatekeeper for lib updates to openstack I want to make sure this doesn\u0027t block updates.  If it goes down too much, starting out as non-voting is an option","accounts_in_message":[],"_revision_number":2},{"id":"5319210f7db3ed03fa35ad3f9efe0401304ad327","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-05 09:13:20.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e how would this help (libs that are covered by this new cross check, citing previous instances of breaking tests downstream could help as well)?\n\u003e I like more testing but want to make sure it\u0027s doing something we are not already doing.\n\u003e \n\u003e how often does this testing go down?  Reqs is kinda the central gatekeeper for lib updates to openstack I want to make sure this doesn\u0027t block updates.  If it goes down too much, starting out as non-voting is an option\n\nHi Matthew,\nThe job is relatively very stable. In fact, we don\u0027t have any exotic reqs there. One thing that failed us recently was alembic that broke taskflow - it seems only Masakari and Octavia (where this particular feature is non-voting) use taskflow\u0027s persistence subsystem which was affected. I was considering testing in here because of that but perhaps it would make more sense to just gate with taskflow instead. Though it remains to verify whether taskflow\u0027s jobs actually test the persistence. WDYT?\nAlso, I am pretty responsive regarding fixing stuff so you shouldn\u0027t worry that Masakari\u0027s job would clog reqs gate. Considering the current state of things, I would also be just fine with not merging this at all. Happy to discuss all our options.","accounts_in_message":[],"_revision_number":2},{"id":"14c4fdb1c41e886a7c813b01bbf7c820b39d1cc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-23 17:31:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"23713f68e4e6976be42c7d7ad11a1b79a4d0b8c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:54:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472a4ee114064dd19cd1387168780faf : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1cd20aac35462182b53ef7dc518906 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed78a368121c40959684867659189628 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/202f4cb2c9a94462ab0c0c96cabd0e72 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3270b0c1972423791fc7329f75321ed : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb7e149efd5749f38aba9c409ef35c26 : SUCCESS in 5m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2493b325ba2e4508a3940b20d6d2fa2d : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a135c7a60d3c4aa29d78f1be56e99db8 : SUCCESS in 8m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6e685e4aba348028c85df873a81c119 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3f3c3b11d1b493b9b6d699fadfbb5fa : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e273d6e8011f4d15b8bda5465b90cab9 : SUCCESS in 9m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/658e196e4a4440c9b4462b5df3178783 : SUCCESS in 13m 02s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/60c788d512d64a68a3d3f6b8099e508d : SUCCESS in 5m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/426121e3fb574952829e744422f3360d : SUCCESS in 29m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8f68b8a3805d49d198813268e632d0d3 : SUCCESS in 14m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6fdd156b9b7410cb674810c945a36cf : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5052efece27d406ca64aba759c8336dc : SUCCESS in 14m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27b8d980feef451d98bf68a5b9c3e7f2 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f88b5d0537a4eb4ac6d3fb0995bc3e1 : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b4438426633b436d83209d7c2913d9b0 : SUCCESS in 17m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff1dfe487ca34f119cfd6df49292e867 : SUCCESS in 57m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/465ecd22463c41ff8da0b508de98943c : SUCCESS in 1h 31m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8dc176f5d9a209d8f61876f948605effd03cccff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-26 04:19:33.000000000","message":"Patch Set 2:\n\nis this still wip?","accounts_in_message":[],"_revision_number":2},{"id":"bf19600f03f6386d4378577db1241247028de64a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-26 06:33:04.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e is this still wip?\n\nNo, I will be glad to promote it. Thanks. 😊","accounts_in_message":[],"_revision_number":2},{"id":"82f04f8190df454f84ec321937c44c7ee5478313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-26 06:33:32.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5b128f4043d1e71b48cc8fff7dd98f01ebabe966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 08:27:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be242086d03f42d596898afbe57b12c1 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46efd92138b04ec29e01d2bfbad8c53d : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/513c9ea9c29f4917b73cf091c63202ff : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b5a5173df5464da879701477aa65ca : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b877e699a05f47f99a096ac0e113ff5e : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/061d8e3dd8694b86af10fcc1b455b7a4 : SUCCESS in 4m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5533cf84e0da4d4a8fc6902812efce60 : SUCCESS in 10m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c07655e24694ae8833ae51e8cd9617c : SUCCESS in 7m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81737778481b407bb5937597995a01b6 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/905c557b5ecb464e870e6e50e9b3689f : SUCCESS in 10m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c3ad0ec4b314c1b8d46f2169f1a6a79 : SUCCESS in 9m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9bd41f329ec0402e8686c51dcd92aa29 : SUCCESS in 18m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/209a252784284b7daa7fbd139cca1c36 : SUCCESS in 5m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/324a11abe16c4609904e205b4a45ea06 : SUCCESS in 20m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b224ffbeccc8492f84abecfadfc260b6 : SUCCESS in 13m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9b8f9eda9c64b83987c8664f90c0da4 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/411c274501ca40618b190d998bc41d6f : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5d3e16907ad4fa7b87738f889a781ec : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db36c66f02464c6f9fbee7f29eebc315 : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db8f835629844ad2a014ba43355665ba : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c413c32e6de24b45a2c722c5bc75274e : SUCCESS in 48m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b81f165df2a46e487c373031ebff6ef : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":3},{"id":"383f68f6044f7f31939257d511ed3e74cbc4fb7c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-26 15:18:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"786e29f62170149d38d45e7af91aca431503b479","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-26 15:19:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"133d8bd4b067600ed2ec696fe06f239d5da4ba68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 15:19:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c351e2002ef773cc9bdd0e15d974a89852ffea7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-26 15:29:00.000000000","message":"Patch Set 3:\n\nTYVM!","accounts_in_message":[],"_revision_number":3},{"id":"1e1cb2cb106113628e46b1925546cf3371b4772d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:05:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffea0fc072f4f8b959d9d3c3d9d8614 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b4d48b8a1e4ed781a93ca656dda049 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7285646ea2504ac682a917bec0aba04b : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/396b84b678384cffb01005ecfefaf713 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d02c7090a1874a53aeab80d54446b83f : SUCCESS in 3m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ee5904479b704a4f8916de74fa08e95f : SUCCESS in 10m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6b527bc48674716a9c5ac864ed96b69 : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a5fd32fb96804164909a4776aa6d9e90 : SUCCESS in 13m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/266017abfb99400dbac64c747452bf49 : SUCCESS in 9m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f7b13ee4e0f04a9388b6fb79df570aca : SUCCESS in 10m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d00fee9236c459bab55ec0e8120e6b8 : SUCCESS in 14m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c30303bb4eff4647ac5860ae8f02cbf3 : SUCCESS in 4m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4aeaafbe0db540e89d3867a6b81d130a : SUCCESS in 37m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c2dca24892f42b0aa319ab7cc4678f9 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78a8088d6c9c4f52aac150bce549f2dc : SUCCESS in 19m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/84a0916f1f6341c9bf2118857a0b4bf3 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7355f372472a48d981e592a0fd4c40e4 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aae7c3769c354fafb2e71826c2012b27 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19f5978d312a438e92bb265c8e3ea594 : SUCCESS in 1h 04m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/258b296a9e314626a5b71860e7b387ee : FAILURE in 1h 41m 03s","accounts_in_message":[],"_revision_number":3},{"id":"64ed56abb2bd49bcd75a2a694986771e093a7315","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-26 19:48:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1c0ba5f653b42fc3d6234c429132f968d5248402","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-26 19:55:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4f070f23076b0f38a1e7a51a7892bc74456b4e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:23:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e102025899244e7bb79c38575450d7b : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7fe6cddcfb4d0d9642e76dfd3ccc89 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be63eba0e56e4fc38c64d7b7f1cb24db : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c0de875a5348dbac0c8f4d7c52441d : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c62e70ac49d4394abb615818bab0fc6 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1545493f17042b194c88977cdc1ff3c : SUCCESS in 4m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/947c13c510104f16a74898532f22480d : SUCCESS in 13m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14b193e963fc49a88cef93bf732c963d : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b786fd38a1d4ddd94c520a42f5a02e7 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78b65608bf154593bec0481fa2952cb2 : SUCCESS in 11m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b440ef0864240358957ef8e7ad6be75 : SUCCESS in 10m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8bc86e7b6f24da7a7ee296c3bc00e37 : SUCCESS in 15m 53s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/adf6b694bf734148bb7cb355c8043991 : SUCCESS in 4m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5740f6f8938741e1a6dd70b5cd479e68 : SUCCESS in 35m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0170a1b780fe480390c081f59b174f69 : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78d88bc2ee45426b8f26673ae66583cd : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0adaf5b0c138494496c4d94f90cfa8f7 : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/53450bcc26df49f69956aca1a16f39f4 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e8feeac594a4db09fc817705296329a : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1a6487fc48d4838a13e94c0c66aa843 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/523be8ba24954609a04d9d586f6605cf : SUCCESS in 49m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d040a7a3502422a8075f6602df29fc6 : SUCCESS in 1h 29m 42s","accounts_in_message":[],"_revision_number":3},{"id":"a8f95609f5dc9285de0bb7203d1f04853c99ccc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 21:24:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86bd5d5495996d305a1c6eadef535cb99951704c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:35:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d7c64a077e43dea46418ed807b0600 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046866a3dda14c56be03e8555ec6516b : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7703792ef40644b28d4246eba77ff0e9 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9284981974044952b85375dfd489abfa : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f978ab2f69fa424bb3dd7c38fd663445 : SUCCESS in 4m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3da0a796bd2643a79469fd24524521d3 : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbe37e77fe774b50b47de8d415f77e07 : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e41a1e745ba64c7a89c186945ad1a29c : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4ba1b1e5b4549aa9b6434208858da29 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f8006b4e0fb94f58a09b71e805e37284 : SUCCESS in 10m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8bc7b1067abc4b0993be186f37c522ba : SUCCESS in 14m 41s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5e7cf72ea4cc4e819159af6a666d5324 : SUCCESS in 4m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/866c5ac8d5fc4a4faafcf048319fa65f : SUCCESS in 29m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46be6154509449fb981bed9993ef8a88 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b650d5f554f7438fb9434ccfd7cbe15f : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8df0072602c74ec9b4a33a9157bdffa2 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb8ec25f209b4e239be26f18bcf190d9 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e561879eec1b4146a4019953d1585658 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab536b5f797a49568c4dc80f02552498 : SUCCESS in 53m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f645b4fa20364669805094da7dc9fa64 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":3},{"id":"f75951ee3d81c5582af3614c2dfa4d14ac138cf1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:35:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"82250245d3486ade87e1a2dcdcf0ce703797d36a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 23:38:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd721d26216c458d830f2650331f54cf : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"51c36c8353c21e2434961ad33a5bfdb8647625b6","revisions":{"b6a1dd57b107155eaf86901a8e825270796fe1b4":{"kind":"REWORK","_number":1,"created":"2021-02-02 14:38:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/773695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/773695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/773695/1"}}},"commit":{"parents":[{"commit":"0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-02 14:37:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-02 14:37:39.000000000","tz":60},"subject":"[WIP] Cross-check with Masakari","message":"[WIP] Cross-check with Masakari\n\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/771437\nChange-Id: I78db12410381908335b86d27d34ec8e7fe594853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6a1dd57b107155eaf86901a8e825270796fe1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6a1dd57b107155eaf86901a8e825270796fe1b4"}]},"branch":"refs/heads/master"},"9ed97fa8fbd753efa9e784758e3b40dc7991c786":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-03 08:07:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/773695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/773695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/773695/2"}}},"commit":{"parents":[{"commit":"646c6644508d43e669275f1fb1ffa500af9b0d04","subject":"update constraint for taskflow to new release 4.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/646c6644508d43e669275f1fb1ffa500af9b0d04"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-02 14:37:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-03 08:07:57.000000000","tz":0},"subject":"[WIP] Cross-check with Masakari","message":"[WIP] Cross-check with Masakari\n\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/771437\nChange-Id: I78db12410381908335b86d27d34ec8e7fe594853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ed97fa8fbd753efa9e784758e3b40dc7991c786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ed97fa8fbd753efa9e784758e3b40dc7991c786"}]},"branch":"refs/heads/master","description":"Rebase"},"51c36c8353c21e2434961ad33a5bfdb8647625b6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-26 06:33:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/773695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/773695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/773695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/773695/3"}}},"commit":{"parents":[{"commit":"646c6644508d43e669275f1fb1ffa500af9b0d04","subject":"update constraint for taskflow to new release 4.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/646c6644508d43e669275f1fb1ffa500af9b0d04"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-02 14:37:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-26 06:33:18.000000000","tz":0},"subject":"Cross-check with Masakari","message":"Cross-check with Masakari\n\nChange-Id: I78db12410381908335b86d27d34ec8e7fe594853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51c36c8353c21e2434961ad33a5bfdb8647625b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51c36c8353c21e2434961ad33a5bfdb8647625b6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
