)]}'
{"id":"openstack%2Frequirements~773779","triplet_id":"openstack%2Frequirements~master~I52189c3792a5302531aa3b3e744a77f34682585d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I52189c3792a5302531aa3b3e744a77f34682585d","subject":"update constraint for oslo.policy to new release 3.6.1","status":"ABANDONED","created":"2021-02-02 17:39:34.000000000","updated":"2021-02-05 05:53:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9bd6018bdc812632b92e384747228cb629b84c52","_number":773779,"virtual_id_number":773779,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-02 23:11:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2021-02-03 15:57:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-02-02 17:55:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 17:55:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-02 18:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 23:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 23:49:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-03 15:57:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-03 16:11:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"89f9a60e17634b9059b406aa0a521fafbb7b74ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-02 17:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c07f431130739ea1f0791ede13ea44ee90f8f2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-02 17:55:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"549d53013d565dab61784e586dd56cc7a433d286","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 18:23:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7c14ec77b4041788d641c362317d03b : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"434a3977bad64930cd0b19dc0af8626546af39e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 23:11:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b8e3adbd084b1dacac629de5a62f74 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4e6bb3a0264ab4a8f04ac3995da7ba : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b71b349d7fe48909e2792e3a42b1a38 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76e59f8a1364e0fa19f4e9dcb31d298 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf1707639eb04934a92827414d3cbab1 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/047793ee78df4bc197d97f5725800af9 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/050d971402204956b3b918d2b7a51ecb : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82419caa9b2c4f09bda081a3d4b90775 : SUCCESS in 12m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f89dbeb17fef46c78e4b7fca3526fbfd : SUCCESS in 11m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a9f8626f8e7c436ebd93d4aa9faddcf2 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4d039749cc9e44e8ae19112a11f00c96 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/05407bf00a9848b2a6ab5b0558964dbc : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08a52243e9c04405bf6f24b98c3e2fdd : SUCCESS in 6m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bffe69727d224877812e29c936fbc063 : SUCCESS in 6m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/745e3216a66047548e6bd40a5ecd313f : SUCCESS in 28m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d09353ac4164ca2b61ebf6490213913 : SUCCESS in 27m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/460d0a604df44932850270681a0ef943 : FAILURE in 18m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8bbea4dc6e7b4a75b2d77a60cfca50c6 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9729b8ef50947239fd0bbad77b03ee7 : FAILURE in 14m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4fa313b781bd4d979aafee7ad1345009 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d688cefff2e24268b08508ec9639a78a : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a5246045098947988c9da372ecf11cf3 : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af8f92d19cc14f3b88564931dd927f70 : SUCCESS in 52m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e35ebd56a18644f3877dd7d2acd60216 : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5ad1d319d6f66dc1188250fabdf6b139005ce85e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-02 23:49:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f345a56e3cda417e56d75a608ddde36a364c8303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 00:22:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/424b1e1ab5984656a5b3c22d830d014f : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e47b3101c7a0bcb95b2464ae40c60c8247c6a950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 02:25:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e693ac2afe4745b2b394ed6320b487c8 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e7e0c687f94099989e906b06fef244 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53bf3795b5484c058ae94a23a691db5c : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73bfe948049542e68637acbfba1a2641 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b3b04eb25f44f40830878884ee7c938 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e0b54aad2fb49e98040d0a732d30557 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b441c10aca94556bfd890906d33e790 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0bc07736ef64e2dae2cd6fff995f2c4 : SUCCESS in 9m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e30f1d193f114479b8d42ae1cb0212e4 : SUCCESS in 10m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4744d1d6e2ac4ad9a43192b67cdbcfdb : SUCCESS in 12m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a141d14c503c44bfbc7c66397c2f2f13 : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d774caae42784011b8ffcaeaf50925b1 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a87f753763a4bf0bdef0cadc9d83135 : SUCCESS in 6m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/88dd3528f4bd46f281732cd50d4dfe3f : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3e66d18c220649c6b0e932522b2b621c : SUCCESS in 13m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4b188900aa7241d5bbe9a500d50b1130 : SUCCESS in 24m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db1cecfc73cd43299ad190312082f5dc : FAILURE in 11m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e4d1981ecb74ee2b94ca61be1ef05d3 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1fdd02f9bfa44c5a169ea31ce5eaa88 : FAILURE in 13m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/479f8d3f67f9400a84fc81d44193e651 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/898da69cd6d3478ba10a61c837906253 : SUCCESS in 6m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44ca46fa5b6749a29074174f4cab1f64 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f532eb5286c5495e83a37bba997a5695 : SUCCESS in 53m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b26e06994a5a4ebebb66c434ec670a05 : SUCCESS in 1h 50m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f8cda2d874fc8aa7296fbf8c6b8e40d4fb141a8d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 15:57:54.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is badly broken in nova at the moment. Sounds like something might have to be blacklisted. I\u0027m investigating.","accounts_in_message":[],"_revision_number":1},{"id":"4688c46714f6ad5a87a116db52605f925e8457e7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-03 16:11:05.000000000","message":"Patch Set 1:\n\nI think we re hitting issue because of https://bugs.launchpad.net/oslo.policy/+bug/1914095\n\ntrying to see if deepcopy the rule will fix it and accordingly we can do new release.","accounts_in_message":[],"_revision_number":1},{"id":"50ad6cc4df62b2201e9f76cb509e1cb32391eb2b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-03 17:59:01.000000000","message":"Patch Set 1:\n\nfixing in https://review.opendev.org/c/openstack/oslo.policy/+/773950  once that is merged, will push new release.","accounts_in_message":[],"_revision_number":1},{"id":"3ea5b27a968b073561587ddea48155ff63c698c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-05 05:53:18.000000000","message":"Patch Set 1:\n\nok, going to abandon this one then","accounts_in_message":[],"_revision_number":1},{"id":"9bd6018bdc812632b92e384747228cb629b84c52","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-05 05:53:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24712d38d91e3599b0e1c03ca6c3d30a20315fc0","revisions":{"24712d38d91e3599b0e1c03ca6c3d30a20315fc0":{"kind":"REWORK","_number":1,"created":"2021-02-02 17:39:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/773779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/773779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/773779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/773779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/773779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/773779/1"}}},"commit":{"parents":[{"commit":"0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c2f0a59fc421a7ef42fe29f7730815bda1b9e3b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-02 17:39:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-02 17:39:31.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.6.1","message":"update constraint for oslo.policy to new release 3.6.1\n\nmeta: version: 3.6.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: Ic312679976481e3082f3523e2c0058b02dd77bf8\nmeta: release:Code-Review+1: Lance Bragstad \u003clbragstad@redhat.com\u003e\nmeta: release:Code-Review+1: Ghanshyam \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Code-Review+1: Oleg Bondarev \u003coleg.bondarev@huawei.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I52189c3792a5302531aa3b3e744a77f34682585d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24712d38d91e3599b0e1c03ca6c3d30a20315fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24712d38d91e3599b0e1c03ca6c3d30a20315fc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
