)]}'
{"id":"openstack%2Frequirements~774041","triplet_id":"openstack%2Frequirements~master~I2d9cf13dc39134aa61305fbf0d961ffafb8037bd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I2d9cf13dc39134aa61305fbf0d961ffafb8037bd","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-02-04 06:22:38.000000000","updated":"2021-02-04 16:49:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3901a19a983922a28f6897c6cbe5c71762fca5b7","_number":774041,"virtual_id_number":774041,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-04 08:23:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 06:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 08:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87699d26618a3116cc39ae8f53ecd5c8bd181cfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-04 06:22:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d4c11298892ebdec425569e0bbbcd6c5f598e44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 06:43:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/53c644ec39a740d0a58c55732af851d9 : FAILURE in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"220bbdbef81de67758421db44a3ddc9486bb6a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 08:23:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaacef86ffbe43e4b36eee4159ade183 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c951db31dc99471aac9b113909be13b2 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56e8ccf4e3d74ca0a0f5a2274d4ab85f : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7771204b5a41efa8406eb59c1acc70 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3aa7777fedc43db9f0a8c35e95a36c1 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fe8404145f746e281ca35224fbc4f4c : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e5626db7b0443bd9db8ae3fa368d0b8 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dac4ed848af24577b29fe814c9e570fa : FAILURE in 4m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c8b3ac7e9fdf496fb4c193be6ea7900b : FAILURE in 4m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ad55725baba44a4ca7157f61438cbe76 : SUCCESS in 11m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2e5cde4a4d445bebd315e7273d32651 : SUCCESS in 6m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/654e3458cfc24c10b3f1cd0594fc6a79 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6982df2b5d9443f7b4b4a55480cf9123 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b01ae71568b344bbb83db5743bcae49b : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/30585f40162c4e5391e7d9cac52f4c5e : FAILURE in 13m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c79040d0dc134fdbaa1929e8a2135247 : SUCCESS in 26m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83c9b054ca15456996462f57de8231d2 : FAILURE in 11m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5305331b938a4285b6e9a55c283069f6 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c23d8dcc34e421b8bc2a079380f051e : FAILURE in 12m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c6f4a1d46b054c2d8602e4f77d6c51d7 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92d9c092408f4885bf57e66d092f4fe7 : SUCCESS in 6m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/524516a2f5c84d97baf61274a88b60cb : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49ec5ed3d3544d24ab61fd2d053a738a : FAILURE in 35m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff3d1bfd2645483ebc94f4a3dd7d146a : SUCCESS in 1h 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3901a19a983922a28f6897c6cbe5c71762fca5b7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-04 16:49:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3955559421103d38d50b68033b17ca5593720174","revisions":{"3955559421103d38d50b68033b17ca5593720174":{"kind":"REWORK","_number":1,"created":"2021-02-04 06:22:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/774041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/774041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/774041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/774041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/774041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/774041/1"}}},"commit":{"parents":[{"commit":"646c6644508d43e669275f1fb1ffa500af9b0d04","subject":"update constraint for taskflow to new release 4.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/646c6644508d43e669275f1fb1ffa500af9b0d04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-04 06:22:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-04 06:22:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2d9cf13dc39134aa61305fbf0d961ffafb8037bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3955559421103d38d50b68033b17ca5593720174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3955559421103d38d50b68033b17ca5593720174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
