)]}'
{"id":"openstack%2Frequirements~774064","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I8718c40cdb95258ea7553ac70c1fd632daf408b7","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I8718c40cdb95258ea7553ac70c1fd632daf408b7","subject":"update constraint for horizon to new release 16.2.1","status":"MERGED","created":"2021-02-04 10:58:19.000000000","updated":"2021-02-05 01:23:45.000000000","submitted":"2021-02-05 01:23:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774064-new-release","meta_rev_id":"9d28b0fb7dd6b2868368541d57c259ad94c11832","_number":774064,"virtual_id_number":774064,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 01:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-04 16:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-04 11:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-04 16:48:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 11:49:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-04 12:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 16:48:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0f0b7b46f98b274731edcfa639cdeb8c21f50a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-04 10:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90988f378a4d1221c4dc6c614509f304db3bbe33","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 11:49:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c6ed3a3434078778a821503b694fdd34d070441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e324222ee4ae44648505f3d7ea3a8516 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5897ba50c104fada773ca0d1c12cc18 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9076dabbd5d4180844af479b59e4ff2 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b24dc4396f845ebb6aea3b5244400a6 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/440bd7420e5d4e58acfac2b7d0ae6081 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b450d0566f7e49269c3c5220269b0da1 : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ea91f78d274748b79e4d5e0cc7b79d2e : SUCCESS in 9m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2295565fab974f21bea960e247304d88 : SUCCESS in 9m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b56c811e839d49008d7d69bff46b12d3 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e40430203a4c49e18c5696ca17fdf4e0 : SUCCESS in 9m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3a13a846136c4636a481a2b17fabc0d7 : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18907042dbbe4f0b8e99316bcb665fbe : SUCCESS in 6m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2868bb8dcb9b4633b7456051fcc7c38f : SUCCESS in 7m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17a58102c5f846ea83c0eb528e82774d : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10a30fd82cb74b8685b30fe87bfa2bfd : SUCCESS in 7m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/633e850cfd724149b64a6f6460f476e9 : SUCCESS in 14m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1dc7eed589141fbb3bb7552ee1acfd4 : SUCCESS in 15m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c080892531444454a2927b4a87c81c8e : SUCCESS in 22m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ed4007c24ba47ee8c2b27fdabc9b3eb : SUCCESS in 24m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/72ca045483ff483e82689c1b62895d65 : SUCCESS in 13m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/41e4cfaf73e940e9be39ecf535890b45 : SUCCESS in 12m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5c52d79cd7e4e7596270600b24ef26b : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cae3655d40fc482199c21efc2f292495 : SUCCESS in 19m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a8363408bc34868af0d87b76afded47 : SUCCESS in 5m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f4ab8df7e40f4f639a6c815bccca4604 : SUCCESS in 9m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0217f249cc244a3d8a0563c7bd2001a9 : SUCCESS in 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e06d05f9ddd49998367197c36428dde : SUCCESS in 57m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/81f764b466b348d78096d072b4030b67 : SUCCESS in 1h 21m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0246750bf83f4d0c8c904683cbb1753b : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"06bbcab1053b4dc5a404ded0d657c3ef4a3fd49f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-04 16:48:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e3ecb60868fe20185658753ed80e41e1ab93cbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 16:49:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66ac37b851dfd1f5bb2083133122c8c8e158e3af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 01:23:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a3e101f6bc4442830d961a02f9b349 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8aa2c0fd910b40ce813b2f711c41c661 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c162b390b24d9098eef56a4ed2189e : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4af8c633e42647658d22842a8ea30d97 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ac420b134d9459188adbed519899637 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f887223c06d410a9720c1ea267fb90c : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7e765940c5334980a1f8310e1c6a7d7e : SUCCESS in 8m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6acb158b45ec4236a4869e0b4ffc6f06 : SUCCESS in 7m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/db01fb3f2e3b4d729bd9ac313429a202 : SUCCESS in 9m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/55b54def08aa473aa94bb2c34c2cc7d6 : SUCCESS in 8m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3440f3ac45a64d839a7965077b83ccd4 : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3604fc6cc5564a06a26930b14fec00ae : SUCCESS in 5m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e96b1e7d9bb24c4a99539fec90970930 : SUCCESS in 10m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e01792319d0444c381967596de4bccbc : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53bad963a28949a0996c0196a488ede8 : SUCCESS in 7m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c9f4c67dd5e64471a9cb81914b3b9268 : SUCCESS in 12m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/72bd3f2a29834d1092a41af08b9523bd : SUCCESS in 14m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4826924b10bc4f7fa577d475bf73de84 : SUCCESS in 22m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9542ad9bb5b49cbb6e1d3e9a901621e : SUCCESS in 23m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c4baf95f1fba41f2bbfa5c70800b2b56 : SUCCESS in 14m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e3373c992b6c4a94ae1e2c392d9d2a33 : SUCCESS in 13m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d111780e2214453d87ee0f4bf0f85503 : SUCCESS in 13m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3deff4dcf784248a5710e7b8d8f0de2 : SUCCESS in 19m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cbdd983432a4a7a821e45cd43250744 : SUCCESS in 5m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e142825f2eeb492abdb8827748eff105 : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/77822acb9e7a44cf85319868c342a028 : SUCCESS in 8m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a2513ad2ca94d0b9d51cb8913a55201 : SUCCESS in 1h 05m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b1b80b5861a34cbf844f1184dedb34eb : SUCCESS in 1h 30m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a6b4875adfbc429eb5b99d6079bc3a7a : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9d28b0fb7dd6b2868368541d57c259ad94c11832","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 01:23:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62467c0570ebca99dfaff2f94bdbe0306a7ba070","revisions":{"62467c0570ebca99dfaff2f94bdbe0306a7ba070":{"kind":"REWORK","_number":1,"created":"2021-02-04 10:58:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/774064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/774064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/774064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/774064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/774064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/774064/1"}}},"commit":{"parents":[{"commit":"af005bcc94496793071344d81eb0abcb483a586c","subject":"Merge \"update constraint for openstacksdk to new release 0.36.5\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af005bcc94496793071344d81eb0abcb483a586c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-04 10:58:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-04 10:58:17.000000000","tz":0},"subject":"update constraint for horizon to new release 16.2.1","message":"update constraint for horizon to new release 16.2.1\n\nmeta: version: 16.2.1\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia2c366658bdd79ec976f90d3fb79fb8362e225d1\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8718c40cdb95258ea7553ac70c1fd632daf408b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62467c0570ebca99dfaff2f94bdbe0306a7ba070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62467c0570ebca99dfaff2f94bdbe0306a7ba070"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
