)]}'
{"id":"openstack%2Frequirements~774074","triplet_id":"openstack%2Frequirements~master~I95f27566440c659d6504da156bdaed69836c3a91","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I95f27566440c659d6504da156bdaed69836c3a91","subject":"update constraint for sushy to new release 3.6.1","status":"MERGED","created":"2021-02-04 12:08:01.000000000","updated":"2021-02-05 02:47:27.000000000","submitted":"2021-02-05 02:45:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774074-new-release","meta_rev_id":"4d5c6942baf31ff5818fb979f025da3af060bcb5","_number":774074,"virtual_id_number":774074,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 02:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-04 22:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-04 12:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-04 22:37:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 12:31:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-04 12:55:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 14:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 16:51:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-04 22:37:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4b84a55b2e0ca902a002aea51bac08d7c7233c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-04 12:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cfb85435b99dae3867652fc36490c0fe0834622","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 12:31:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cf327bfcfbf76557261845f67b06de1dee7fd0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 12:55:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/167faddf098d438d821e8e50bb4ba15e : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0fd7c8666376314f52f109bdace5168203e3a548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 14:40:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ca5a24111d473dbd32672a07b7adc7 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c49d9e6d0f41dbafb3e015df7ad1d8 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3d5650b98d54aab991f777be839f827 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0fdae6aff764ad68c7047a8f5d08c2f : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10424206e61144df86d57ca4ee1571e7 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5e790e8d75d4e528890b3beefe049a5 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9029eaa4ca2c43a39acb1d4cc63c3882 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5d83243bc746472c890d8c466974c9d0 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c306358d26924b86a18e3e707a361cf6 : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d4c5e36397314fbb8f22f8f937f90228 : SUCCESS in 10m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3aa60023e674de3874f62b7ea2634bc : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4703fd23da3244678c0177acd5370536 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb86257396dc4cfc8547c8a68e8277c6 : SUCCESS in 7m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cdc848e6c7794523a6a2fd1504d682d7 : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af08734e0c354ce395ee1533207d293d : SUCCESS in 13m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/80c24ef34d7b4c2da3a1ad752e8c61a5 : SUCCESS in 26m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d7e942c8d1644d9afaf0122263050d8 : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de551bdd30a24e6fb6ff6b570d36b6a3 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/504a332da0e44376b206e764a3898204 : SUCCESS in 14m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6887511997ca46a693ccee50659ce43b : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08a452650b464822b17c63cb0da26d4c : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dba28c8f63624beebb219ccb45ddf11f : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0ec0ba16fb249268606912bf3ed333a : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5403880b62c244db882b682aeccdbd01 : FAILURE in 1h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8d7d9f0ab1739f769eaa0c39d77165b46c63a5e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-04 16:51:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"159e39d6aad0d9c1d9bdacd9432d1e03ac681dae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 17:34:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f5e90e50d0204c9eb905eb872689772b : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d201c5f7624637378d3ac98dc12e092a6c491ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 21:56:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c32fa845ee08410d9b8ee1b0c06d33ac : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c1f74f110142d7a14d008914bda833 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e28c596e9ea432a957f3ae488a551b4 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cc93973c4bb462ea509ce9dd9b36dbc : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252aca562b0e40faa0bcc3fa44978fce : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f8d9641ad76439b94b6086eb5f55d05 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f02ebe0b23f94342af0e3fedfb26c03e : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa72e540430146ee8ce9c4d9d7e9987b : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03975668a198424f8b81157513ecddf7 : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d764d05673cc4a47992e5f54003d036b : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d55b8adced644fdfaae5218afb01c81d : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cf94ecf9e42f462dabb3185673963dba : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0079461ee7a347178b4d3d9b59647eb1 : SUCCESS in 6m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d74a93ce38144e359eae31056c2b9d4b : SUCCESS in 6m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/35cb5f9829524c718797f6e5d7ea0588 : SUCCESS in 14m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/accaa447a6df4f12bafd7977ee1108fc : SUCCESS in 25m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac6528f1cf294abc847cf749dedfcbf2 : SUCCESS in 11m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51d103caf16740fe83e15e846cbe61f1 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50e50d1913644cd4862537534efc2ec8 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/41711fa942b94e9ab63da4aedb5c6742 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a19f929fb31f496d80bd61e18a27263f : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/246e2ef654a24548b2392747af56c2ae : SUCCESS in 22m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e575d7265ffc4dc4aaee7fded1140521 : SUCCESS in 1h 06m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6995f2befbd94a798a23447e782f4614 : SUCCESS in 1h 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c4b282781d7802fcc680a8c6c508dccc9deac81b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-04 22:37:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef26c9f2048852f23c262879bb3684868ddf68f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 22:38:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e49d335394f6a0473b17749ab6b522d938319bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:45:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f72d9c8fa347388d211bf63bd033fa : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290166238c5c41dca01fc7cfae9fd81d : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f47ebf96f92440e82e1c0f1f4cb2d43 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fd0f7afb3b74adea3830def1e348c7d : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b629cd5cf01a46bab2346d6ff8504aff : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06805c4230fc462d98d5c3dc9566af3f : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95e7138f753140d5a4a45f4217d4e8f0 : SUCCESS in 9m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e757e734642a436cb2b9fb31285f0489 : SUCCESS in 13m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78561c54eee54e949ea6588e0e371db8 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/172d153e87e545a28c0be1a815d47ab3 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fce53a686c5e4170924d1c8e513bd660 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/592c973f424a4067b5f5eb7486503848 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a76e36075c4547c88a994a35cc33136e : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6fbce957a81f4cfead253635cdb39096 : SUCCESS in 13m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bb1d14d1debf4f1fa0775d1425b31a8c : SUCCESS in 25m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0d582d4903bb4e9ab725c14f404a9ed5 : SUCCESS in 15m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04e3461e4301401ab0d4d65b6cb85040 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19d877c3d59848b1b5ed672a1d4ff270 : SUCCESS in 18m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a87df7b420a04ff3843796cc9883603c : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca1c2331fe824844bb132fafdad41bdd : SUCCESS in 6m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/adbfc2dfb1b24f67823d4a69b94a8713 : SUCCESS in 16m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ace7580948e8439b8b803b2ace813b6f : SUCCESS in 55m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70088f7fa8894ee28fe0b9fd98aec1ca : SUCCESS in 1h 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b04be9019ad7bf2fbeeaf388a44aae978466ad21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:45:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4d5c6942baf31ff5818fb979f025da3af060bcb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 02:47:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd7142a838c4cf9825e881f99c5d622 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b4c67254e2175d2e7b9cf0114f11fbc690606ea","revisions":{"2b4c67254e2175d2e7b9cf0114f11fbc690606ea":{"kind":"REWORK","_number":1,"created":"2021-02-04 12:08:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/774074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/774074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/774074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/774074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/774074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/774074/1"}}},"commit":{"parents":[{"commit":"646c6644508d43e669275f1fb1ffa500af9b0d04","subject":"update constraint for taskflow to new release 4.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/646c6644508d43e669275f1fb1ffa500af9b0d04"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-04 12:07:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-04 12:07:54.000000000","tz":0},"subject":"update constraint for sushy to new release 3.6.1","message":"update constraint for sushy to new release 3.6.1\n\nmeta: version: 3.6.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Bill Dodd \u003cbilldodd@gmail.com\u003e\nmeta: release:Commit: Bill Dodd \u003cbilldodd@gmail.com\u003e\nmeta: release:Change-Id: I3a9da98c363fc9322d6f13ce600be1ebfd5b7ed6\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+1: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I95f27566440c659d6504da156bdaed69836c3a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b4c67254e2175d2e7b9cf0114f11fbc690606ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b4c67254e2175d2e7b9cf0114f11fbc690606ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
