)]}'
{"id":"openstack%2Frequirements~774290","triplet_id":"openstack%2Frequirements~master~I9bdc41ddb839941074a549675df6527d2e40dcae","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9bdc41ddb839941074a549675df6527d2e40dcae","subject":"update constraint for oslo.policy to new release 3.6.2","status":"MERGED","created":"2021-02-05 17:26:12.000000000","updated":"2021-02-11 23:54:21.000000000","submitted":"2021-02-11 23:52:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774290-new-release","meta_rev_id":"e98f40f87fcf75e7e9457bb51a05cf1673d27fd3","_number":774290,"virtual_id_number":774290,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 23:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-11 05:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-02-11 07:04:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-02-11 15:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-11 08:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-02-11 14:35:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-02-11 09:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-11 09:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-02-11 15:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-02-11 14:46:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-02-11 15:26:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true}],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 18:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-05 19:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-06 00:40:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-08 08:40:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-02-08 17:07:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-02-09 08:51:57.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-02-09 09:36:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-02-09 10:06:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2021-02-09 14:04:51.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-02-09 14:04:51.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2021-02-11 05:09:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-02-11 08:31:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-02-11 09:02:34.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-02-11 09:14:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-11 14:35:46.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-02-11 14:46:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-11 15:26:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-11 15:27:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"08138ef846137f1820977099dc75a93e98ce32e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-05 17:26:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c479d3d971c7741d00b2296b2f5dc991412da843","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:02:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e95a68beab4e4c56bf05f5830694ba1a : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0ed5aa63424f12292fcce0f1a15902a1c8cc3a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 19:41:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ac4bcf892d4721be34213690343bea : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e2b583e7454e2b8db474fc679244ae : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80dc27902a85485aaadaac78285cc6eb : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e8dc6338d3140f780826b3cb0682cfd : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e492ad758e241768c63929ce1433574 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6691f99c10f6484981f8357404e71109 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6006ab62026246d0aa3c18647ed786ce : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28a5fbc0bd2946a9bdb97c47db856bda : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0cf170e5d7054c9aaa97af6fa09394cb : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e8c79fb791349caa11fb52fad2ffbef : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c692e7d909f24e1b8f885ec1374365aa : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/28206a1950bc40f9a282ddfaa5fa02d6 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5facabe85adc476c82f47d90199a01aa : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/481d560067a2406db6b332d2d00c4777 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f11b19329134f0fa99dc8f35db065b1 : SUCCESS in 14m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f4a903fe4b740a1a9b1ed87249f8fa7 : FAILURE in 14m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d4dcebdca924613916cc4d88eb228b0 : SUCCESS in 12m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2844737d539d480f98524f00e6b95cf7 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/518e9407df18491392671c875f3bdb65 : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4052e11ea6de411a8560b501f9c50f5b : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65604ab732a24545b9984f2968b8b640 : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/570aceac0bc54a5f83167526c07f21c2 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffc2461b8efd424aabb553ee59770aa7 : SUCCESS in 54m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae7f2006380f45f6a40956963fc5d5d2 : FAILURE in 1h 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fa3cc565f8893dfb18e6e5e9d9065b90e5e4e26a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-06 00:40:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b3be4dea2ba59731a4121f9ba34de553e920bf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 01:12:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bde6339d92c94970a9adfdcf5ed8a074 : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":1},{"id":"12f04f4cbc267dcc01cb42b2317a572ff7c69120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 02:03:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ecec0bb7854e378eeb23649b92d7b3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcede27cf40c4bf29f5eda7361d55518 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76e0a59bd33a455d8a52219549ff6c2a : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d851b6185234c828c6cc114d6f9d38c : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc67ede57ebf4a508a54aeb18a505dfe : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/582560e0e1db461dbc924d56cb219642 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7197cd4637d44245922dc4925755e9c7 : SUCCESS in 3m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e94c57d323b4107b9179cefe570ce5d : SUCCESS in 7m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a3bd0c24ee96402dbf89c8aad1addf6e : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f7772b26be748d983040f8121566e75 : SUCCESS in 8m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e89df33bcb144b183dc6980ac64832d : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a78fc94976d24f10a3359d2583a43237 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea538330ba0b4eeeb6d98329ea7eeae0 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e2126d75415d4f2b80b402c92fd42cd9 : SUCCESS in 5m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4dde6c6c7f28430db74a3d3ab2ed810f : SUCCESS in 12m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f59ff01d875f4a6eb822e86687861467 : FAILURE in 11m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d2ce0d325804405bdbaf76d341b2845 : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8985188b12ea43c89ee096c1b2423164 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d343459bac004ec4a01b16ef2aea133d : SUCCESS in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/da6a253d2f854c1997756e1544d4df8a : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93479a8db4b6420fbe7caa533f4b3a1f : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e11d22912644c64ad67d52542873884 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbf0833c5e85400ca0eb96434765772a : SUCCESS in 55m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a07ddd3ac777449196a48577ccf93d43 : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"36676840cb890e0f222414effb35af90f185e0bb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-06 05:10:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c18c608886ee040c928ae6530a49a1c3d155be0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 05:42:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aa824fe4d2e048dfb1963ad96549d89e : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9fdf71105e6b5f13a341d45683efb8ccf38a303b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 06:52:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0257f7e8f5d84559b6f7ac531b847522 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08744a0550e84fe9b920c4a737a18f0f : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cdf0fa435ab4a1c8e52e71925f84335 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9261064b9792432b82d34af6917b5298 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c515184160c4085a81952cee6456148 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d48800ee88c24bebad968442c8a16ceb : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78c0a5c948214a7bb2692b09e9a917ce : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d74c254dc2a5458b960e6323135ee7fa : SUCCESS in 8m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/def26bcd7fd24b078ac90ee154597cf7 : SUCCESS in 8m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20594ff2e0554dbd9ad43e14e08e23dd : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1360a82449364b1a90be3e8909fd221a : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0450b1380bee490caec46286738917f1 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7db5baa2d5a74053b63a5dee31d2d373 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2f8c9cac1cbc46d9b8d5bfe073e468ac : SUCCESS in 7m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/080682c6993b435fbce25bdcdc9f1337 : SUCCESS in 15m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8348c761e634d40b8c4da7de22386b8 : FAILURE in 11m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/72e037dc3ad9435cabf3f0995deb7eb9 : SUCCESS in 11m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6df8904970440318ceca7b7900a1b9b : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9a882bf03ae439b9d77188e50d6ae22 : SUCCESS in 15m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/940e44955fdc415c80bb9c97850303ba : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c519778718f4386a48b758167cc8ef6 : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3285ca22b2514ac89bc62fd8cb52ab2a : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98eb1e57b5664976a899163a3591f6e2 : SUCCESS in 44m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25f8823c8b4941edb4a8eea129dc960e : SUCCESS in 1h 40m 10s","accounts_in_message":[],"_revision_number":1},{"id":"74e4ef91cc8b4d2ab080218dda1959060e6a73b2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-08 08:40:39.000000000","message":"Patch Set 1: Code-Review-1\n\nWe can\u0027t promote this version, is failing in Neutron.","accounts_in_message":[],"_revision_number":1},{"id":"f49bad07e7cdb49795ccd5e1cb1a3bcba65a8376","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-08 17:07:49.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e We can\u0027t promote this version, is failing in Neutron.\n\nhumm, I cannot reproduce it locally or even in https://review.opendev.org/c/openstack/oslo.policy/+/773947","accounts_in_message":[],"_revision_number":1},{"id":"0a1292dc6d13442aa65a867c09598bfc0a29d178","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-09 08:51:57.000000000","message":"Patch Set 1:\n\nCould it be because https://review.opendev.org/c/openstack/oslo.policy/+/773947 is with python 3.8?","accounts_in_message":[],"_revision_number":1},{"id":"cd082b49d8b8e35eaf4ef57d0cdd69cf4c80d490","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-09 09:36:56.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e We can\u0027t promote this version, is failing in Neutron.\n\u003e \n\u003e humm, I cannot reproduce it locally or even in https://review.opendev.org/c/openstack/oslo.policy/+/773947\n\nI think we are hitting this bug https://bugs.python.org/issue10076 which was fixed in python 3.7+","accounts_in_message":[],"_revision_number":1},{"id":"d77aa7db701b8035bb0c58422eb6507736b2f5bd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-09 10:06:09.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Could it be because https://review.opendev.org/c/openstack/oslo.policy/+/773947 is with python 3.8?\n\nYes, that\u0027s the issue. Neutron has a custom policy check type, FieldCheck, that stores a \u0027regex\u0027 attribute [1]. It\u0027s not possible to deepcopy regex and match objects before Python 3.7 [2][3]. I\u0027m guessing the gate runs on Python 3.8.\n\nQuestion is how to resolve this now. Would a fix in neutron be acceptable? I think it\u0027s reasonable, personally.\n\n[1] https://github.com/openstack/neutron/blob/ab8a17c6a49/neutron/policy.py#L377\n[2] https://docs.python.org/3/whatsnew/3.7.html#re\n[3] https://stackoverflow.com/a/52885510/613428","accounts_in_message":[],"_revision_number":1},{"id":"381f0576b8b300ff0eb28f139c6d4775576e137d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-09 14:02:24.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Could it be because https://review.opendev.org/c/openstack/oslo.policy/+/773947 is with python 3.8?\n\u003e \n\u003e Yes, that\u0027s the issue. Neutron has a custom policy check type, FieldCheck, that stores a \u0027regex\u0027 attribute [1]. It\u0027s not possible to deepcopy regex and match objects before Python 3.7 [2][3]. I\u0027m guessing the gate runs on Python 3.8.\n\u003e \n\u003e Question is how to resolve this now. Would a fix in neutron be acceptable? I think it\u0027s reasonable, personally.\n\u003e \n\u003e [1] https://github.com/openstack/neutron/blob/ab8a17c6a49/neutron/policy.py#L377\n\u003e [2] https://docs.python.org/3/whatsnew/3.7.html#re\n\u003e [3] https://stackoverflow.com/a/52885510/613428\n\nI agree to ix on neutron side, or I will suggest to move FieldCheck on oslo side but that is separate thing.","accounts_in_message":[],"_revision_number":1},{"id":"910a2e1fea54262f93e1a2ca9ace862cb191fe61","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-10 14:50:44.000000000","message":"Patch Set 1:\n\nSo, should be good to once https://review.opendev.org/c/openstack/neutron/+/774621 merges (after https://review.opendev.org/c/openstack/tempest/+/774766 gate fix)","accounts_in_message":[],"_revision_number":1},{"id":"5ca987ee3e89dc610f4e0bacca8f70d6e44b01f9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 03:33:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b7a44721d7cd1589afa28817feed80fb5bd827f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 04:05:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/62691bb686bf4013a6e1f7007e3e268c : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"446d5a570f5cc0a44d1e60b28e29d8dbf233245a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 04:53:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f2f97674354bd2ba83fae8e2d7eed7 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3c789ff8164f438eebb5fcf799119c : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f223c785ec354e08a48442b63b3221bd : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ecf2a38e2eb4c8a8de5b3ca9a049e7d : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e8f790778ae45e2ae8ca2f39f50271c : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d308c42658304e56b1765dcd37dc868b : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d36e092df23d453ba5d34d3247baad4d : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e3e8f5602054644b3832d73e91e9563 : SUCCESS in 8m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b6695872aa54b5e9b2c359742aa7c39 : SUCCESS in 8m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0591c3349ca647eeafaf554e983d63a9 : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f8294d5a85c4003ba35b723a2a3ae82 : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d19f8d620994514b0e8865e71b83acd : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11f74cd33f414d1093e87c62bcf944d6 : SUCCESS in 6m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0273cf73ee0f45f7bc4b0a1a09abfc44 : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e5d7fb630e414fa4a8521ac83e41251f : SUCCESS in 16m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/532ee1f38799429988ba73e68c247304 : SUCCESS in 27m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/896d735d99e445c59334b1f8f7888577 : SUCCESS in 11m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c17545470d8d4a888185dbd010bd1673 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e811f787701949c9bad0028cde3bc3e5 : SUCCESS in 16m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/137b76f19c814d5099812e6624cdcc1e : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12579240d0e44c359c0f8e2b1afe5a6b : SUCCESS in 5m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e3cef4f5d3741328375d9fc14cb8719 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad4fba05d789416ca1b55901b41ec4e9 : SUCCESS in 57m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6c51f4f3e8440c1a428e242428b13a2 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3265bdd586aa67cfe11934f0d4744f89fa0bb81f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-11 05:09:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1aff81dfbeb1b56f4f0dd91a0ff241a9a128c1bf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-11 07:04:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a74c1f3345ed6fd262847435a05820304a12b057","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-02-11 08:31:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e7cefaf1293001438f77ba7e84e050ea97660543","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-11 09:02:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"abf8c77d9d1f8f6c73869dd24c2579211b89e4c6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-11 09:14:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ce4727df6d0c8fc29595b3bcaa664538cbfcc82","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 14:35:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8a7921e109d80a6eb0a2b6cad1a815302fc7c154","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-11 14:46:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf4fc324ce9a4f0fd2c62f5d8ff4f0b1e1f5b584","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 15:26:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9dc1a12b47937c7c491afcc22651691ff20f13a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 15:26:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d84b4221a7f806db1a59c807cdc82c926f963460","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-11 15:27:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8f65ca30b75653929e55cf3bb75f9d0c1d53bee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:09:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445797aab03143e5a62f552241e49718 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b6f06b846b4431b18b07d43b74ec2f : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b6cb1a6bc5b42f49abf005f79d58b78 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1fcebdc28749079b508a8088c501a5 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a3dc028617446af920ec565efca2e96 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e59e70856450438aab93c8434a1b5dfc : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/94ab927a962347abafbe97134fed48ea : SUCCESS in 10m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e66ee4c9093477c9308222118b9342e : SUCCESS in 11m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/60587d888ace4b5b813f417faa556e77 : POST_FAILURE in 24m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff7480a45bf74c50a7ad0f582cbdd0d9 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/da1b9e7cdef34988a97130cc85f7755b : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ed0b4c20f744639a2c5be1d95dbcc0b : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1fe4726a16804ae596dbf716228555ad : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/994c579d745b4bbe9a43a658e29b8b90 : SUCCESS in 13m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/954ef3cb8414466b8c5dd797513a1a0b : SUCCESS in 27m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b395b5ad04e4813a081653e01ee680a : SUCCESS in 11m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20baa3de593b4408bd0db5597396c9d0 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63a2b02776174c6e936d51bf94c8e501 : SUCCESS in 21m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/14013964e5b345cb807213ab96101055 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92e2a7aee7fe41ab8cedfb1c7c9f04bd : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c68fe5bebe304f4b9f7a7f005e02adec : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c499a851bebd4f2999ef4bee5774a7fc : SUCCESS in 58m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4132f7e1f207464da1fd6189414a7096 : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a9744a8482484f77950fc36c95bcb1f4e649bd5d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 17:14:54.000000000","message":"Patch Set 1:\n\nrecheck due to a post failure caused by storage issues on the infra provider","accounts_in_message":[],"_revision_number":1},{"id":"137d5069feae8247d0e98560fed280c6fe965396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:50:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/da723daef4024d5bb0647584207a4702 : FAILURE in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"940b46777e95ac5c4f662f0cc6504a61cfc72da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:26:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ea598bb2f74ce490e3c52bbed5ce20 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9098672340d14a8390d7e6833b70feca : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8681f9aff9b46b1aa1b95da2f383f12 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2774d335252c4421a22d211598604f49 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd91ab19acee49d68d9d6cfbfd662626 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81dad3538baa4c728bcaad0a05c1db6c : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd827f1241314231b67e28318430061a : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/baad1183d4734fde92af3af84f9c9303 : SUCCESS in 12m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d85fdb17d1e44c1e9c48419e8ff59729 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ec9f088305944ad5bb4ba49345a49c81 : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8837682b7e004dc490b29e347b61d2d8 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/19f07d5e660f4931be5fecc3fec7bb54 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ff5c1216a8a4548ac45c62f03633e51 : SUCCESS in 14m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c2d886642997408288b9061b3600da26 : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af1d2666e26643e291e85bb59929ab66 : SUCCESS in 12m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8e097e1125c477d92a10beddef3c13b : SUCCESS in 29m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/21bae0ba7c09498f840279eae3b1f0df : SUCCESS in 12m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c646e1a3d965496ab7b26abad670e7d2 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6d781383c184e95bd8308c3d7160fa5 : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/03b5c089e4734cc9b60a61a01987559b : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd272d0fe70c4c6f950de5e362db05a3 : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/30f9472829ae46ecb0dc9cbde5792a66 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c44d585c6c04d20af8a54ac94c6208d : SUCCESS in 57m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27a5083c4a4b45a8be8816121908bfb6 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5c1f958889a5e59831354b3db63f1c8ea57cb682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:27:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76f58c0be0a4dd9311c4f86d163c1b1fe63e42b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:52:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae467a0b6b054d3f94d7a427cd0628cc : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c75cdd8b995042f999abbb3e38278675 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f53460719d574f7c88a490704148bdef : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e94842a0b88645b4a74516e2757b2f01 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c9afa7cd1ad40e0a906ea05a74b6445 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/221373cf6f494ae88e216e22520cf12f : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/879e5bc138ce4770a87985ebd5c442e5 : SUCCESS in 11m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ca2a66973f94e7995a8f64f03b6f809 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/083f940313ec497f8e6563a5236cfad5 : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/19e769454ced4e13bb90d5c2b000a71c : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f631e5cff734078b011c6968aacf6c6 : SUCCESS in 7m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6264b425b8745c38770e8d49f3bd78b : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b39c0affeee64425b3b325978889c0a6 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7a87da956cc44188294d703898e60ff : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c4ec1e3665994ca0a03825679fa76679 : SUCCESS in 35m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8b80a184a3fc4845b1866bbf14f38631 : SUCCESS in 10m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b0f9b34d82b4122ba90d84219d15b8b : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fec4224f47fa44d3a96be1f62218be73 : SUCCESS in 19m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/40ed1948f7224425bbf03d27cb88fcb6 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f43f9b7b2800447c87b2fad2658e86d8 : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2eab2d359d8b4855a26f5eb9a535b118 : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e59b4fba4904f6d8df3fd9e26a59bad : SUCCESS in 51m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3075fd6983994eff8f075ab336e0994e : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"19f1855e284ee60e4a8cd0e5c5db9cc90e4198d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:52:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e98f40f87fcf75e7e9457bb51a05cf1673d27fd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 23:54:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/223ff3ae89e14b5981b68e1d16167a6f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f380e087763f30680c0688ec43a530073967e9a5","revisions":{"f380e087763f30680c0688ec43a530073967e9a5":{"kind":"REWORK","_number":1,"created":"2021-02-05 17:26:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/774290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/774290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/774290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/774290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/774290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/774290/1"}}},"commit":{"parents":[{"commit":"cc40d6ef13888e86ed36c70058350cf20eeda1fc","subject":"Add python-cyborgclient in openstack/requirements.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc40d6ef13888e86ed36c70058350cf20eeda1fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-05 17:26:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-05 17:26:09.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.6.2","message":"update constraint for oslo.policy to new release 3.6.2\n\nmeta: version: 3.6.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Lance Bragstad \u003clbragstad@gmail.com\u003e\nmeta: release:Commit: Lance Bragstad \u003clbragstad@gmail.com\u003e\nmeta: release:Change-Id: I73528c3ca8a9f3b9fcec6326129d04efa78615fb\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9bdc41ddb839941074a549675df6527d2e40dcae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f380e087763f30680c0688ec43a530073967e9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f380e087763f30680c0688ec43a530073967e9a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
