)]}'
{"id":"openstack%2Frequirements~774347","triplet_id":"openstack%2Frequirements~master~I5cf25fcba050b07cf0818278484636f5715a424d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I5cf25fcba050b07cf0818278484636f5715a424d","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-02-06 06:19:39.000000000","updated":"2021-02-06 09:34:03.000000000","submitted":"2021-02-06 09:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774347-openstack/requirements/constraints/noclob","meta_rev_id":"465629a088e6f3932ace97d0a102618935ba4312","_number":774347,"virtual_id_number":774347,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-06 09:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-06 06:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-06 06:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 06:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-06 06:40:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-02-06 08:12:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8892fb84b1b418f3462ea70f1747afcfb6543465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-06 06:19:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd6f7e3151944b99cc02f9bafd96df0c37c7b680","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 06:32:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c49fa59719c741ef9faca3c53751d2e0 : FAILURE in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6019af7f8f636aa49b1c19b343d8d97107a24df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-06 06:40:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72ace545476726bfb088802ec83590657a081dfb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-06 06:40:37.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"72eae8e93d1e1bf3f1fa06ad25d5d1846e2140ee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-06 06:40:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6566e2e04f50627f84e1f5a9cf2f5cabf95b07d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 07:13:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/376ef12f46e04f52aa630192eabd8150 : FAILURE in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"23cb10d770e7e8cb0d28e86e4bc5294df7e8e8b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 08:12:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260dc909ad784ee3bc92f82f00beaff3 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2cd4cd8bb743fab568f8671a4b60d3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5b3d1e35f6437baaf570a45722a00e : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c73c2b055ab5442494b909ef73c65b37 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc5aa8402da64f35828f25c13f120060 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/822de142fae84e42bec0c13507a79cb4 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf67ee6add8e4b7f89fab9637107370a : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f8ad2a8ca4c489a80746d0d907344a2 : SUCCESS in 7m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24dc6a0d2ec8408a9aab1beaa1bdff5f : SUCCESS in 7m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/471e95453cc942a281d544c7d8b7c739 : SUCCESS in 7m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c04ae72b6fc343deb1c1ff180152669d : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/47aa7c9d277847faa720196407fdced4 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87e6e90dc8374026930689ee0bc2c6c1 : SUCCESS in 6m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7a05e9a91ac742b8bf3946f7e3fd49e3 : SUCCESS in 6m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/791aa3bdc6d54c4ebd0869d22d81f026 : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ffe331ecec4d4f6ba8080cddff5c1095 : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce929faad806465581c7260c419d710a : SUCCESS in 14m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b1c5aabff2b0430b90894f53f9df917a : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/587edb381fe44c2180a32214545a571c : SUCCESS in 17m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df0197f0a24247a9a380673a9fc44876 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83bca4f07d20459e96511ec8556863af : SUCCESS in 6m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/775582e0738e4bc495f4ef80223ac44e : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3cc4233fda54966a4b9a65ad771de25 : SUCCESS in 53m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31c6a2ff36ff408280124dd5703ab83a : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":2},{"id":"28d8842a9f1abe96b6799a5c7180f369a82143ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 08:12:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d401218020bef6771a02a875a8440580d0673452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 09:32:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0edbf4586204a65a615b2ada4d8fe88 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d991c3589dcf4723a1c7086dd11d9dfa : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fafd58888298433195ce399b3e689e08 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/330ba81a6bf54ac2adb0df5407046bf0 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acb2cce10b0c43fb84641bfddcd1029f : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0b410f4f7434c078c0c6f71047e6281 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a34ea8c994d74ca4952440b16c18c545 : SUCCESS in 9m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d9f6a2e471648c4aee903098446bca6 : SUCCESS in 7m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b598bad90f5f456b8bc012dc4662d248 : SUCCESS in 7m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e3dc6913045d47e79352986f23a74eaf : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8bd428aeecfa4c219e7bf3798cb67cc8 : SUCCESS in 6m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f4c7736368f48748de47f76fe2db7ad : SUCCESS in 7m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4987811ec808484dbb18935194061650 : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/88a9aa7516e2440f98227270b0cc682d : SUCCESS in 12m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c789e59d8f60400caf66b825c0eb5f4d : SUCCESS in 18m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/33340b7e502f4823bc60a96f7d94d5b8 : SUCCESS in 10m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4bab77d6b0694bf4af810f37dee56696 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b7436f91fa34cf5b5e777511049b5cf : SUCCESS in 16m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f1eeabfdb2e949bda23c072e7b4ca6e3 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f3a9b93f9eb0497a9d7b0ee4fbd625e5 : SUCCESS in 5m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74f17a0d62484cb796d69b2cd2e2b2fa : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9b484fa87ec49eba98bfa6503f5a386 : SUCCESS in 39m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7642be3810a47659e3c88ea5ebe39fa : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6e0a8169c3df3b2a3c5c470602ce0fb34dac85b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 09:32:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"465629a088e6f3932ace97d0a102618935ba4312","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 09:34:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563be7d7ef234b4abdfbf205774d1960 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32a1a608513986a0c346eea125b92552feb87296","revisions":{"6031b995cc31ba68d92bac8ab07ea9ba8e98e6fc":{"kind":"REWORK","_number":1,"created":"2021-02-06 06:19:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/774347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/774347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/774347/1"}}},"commit":{"parents":[{"commit":"1f03c0a68cf43ee01b781dcc738440baf3ac85df","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f03c0a68cf43ee01b781dcc738440baf3ac85df"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-06 06:19:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-06 06:19:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5cf25fcba050b07cf0818278484636f5715a424d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6031b995cc31ba68d92bac8ab07ea9ba8e98e6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6031b995cc31ba68d92bac8ab07ea9ba8e98e6fc"}]},"branch":"refs/heads/master"},"32a1a608513986a0c346eea125b92552feb87296":{"kind":"REWORK","_number":2,"created":"2021-02-06 06:40:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/47/774347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/774347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/774347/2"}}},"commit":{"parents":[{"commit":"1f03c0a68cf43ee01b781dcc738440baf3ac85df","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f03c0a68cf43ee01b781dcc738440baf3ac85df"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-06 06:19:36.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-06 06:40:28.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5cf25fcba050b07cf0818278484636f5715a424d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32a1a608513986a0c346eea125b92552feb87296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32a1a608513986a0c346eea125b92552feb87296"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
