)]}'
{"id":"openstack%2Frequirements~774446","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I37a4d48f621499b2de238cb9b22afb0528772d74","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I37a4d48f621499b2de238cb9b22afb0528772d74","subject":"update constraint for oslo.serialization to new release 3.1.2","status":"MERGED","created":"2021-02-08 13:21:30.000000000","updated":"2021-02-08 16:45:09.000000000","submitted":"2021-02-08 16:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774446-new-release","meta_rev_id":"6b671091c2f00e303049c3519919160e1e66b920","_number":774446,"virtual_id_number":774446,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 16:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 13:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-08 13:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 13:26:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 13:25:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-08 13:26:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-08 15:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28873a97d8959f5ddc85216f01316452e0a55377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 13:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6410268b9e1bfd4e5aa115580e275bae03a45785","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-08 13:25:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ace0b9dec6068b3ec9cbb4a10882f6e80dfa8ab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 13:26:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"898eeb0a3cd2905f85e0aa510d3109fe2f76d7d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:09:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e64451a554af416eb42b2a9f6f99cbaa : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75f5f415e02948318b1eb355f0977a49 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af70b5d2b1bb422caac9a50ea2f96926 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ee368abf4cd4435ac259dd5d73a632e : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dab6645d90134dfc98d929b495665785 : SUCCESS in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/60f7a20ac448479ea235c7a29b2ef8b1 : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c0e6362aea0d4fa5af85e055eee78cda : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/af761a4cf0b246fcbce22d335e565668 : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b01fe8cfd4dd49db9657bf1bc44c0894 : SUCCESS in 7m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9f72769d71c49158ffaaa6e4b212f8c : SUCCESS in 6m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74766e22e0a34843aa225ec1d8d32f73 : SUCCESS in 20m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8345fdf37154ef6998b9ec0e726ac1d : SUCCESS in 26m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/acd99672dc89456d863ff99be6cf21e8 : SUCCESS in 14m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/95078144ad2745e49816094573c90693 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9999ef560504fd89bfbe18de871cdd0 : SUCCESS in 22m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c22c276b39c8422c9b8f59080a8d8c2a : SUCCESS in 5m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a3669c2153b943fabb055e57141be195 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cd8ed96fc3f45849e033e720c7581f4 : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ee6863e3e55c490fb4ca4b5b29c5695b : SUCCESS in 9m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e946e568b38347549e2eb311fec02800 : SUCCESS in 9m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5dc68c4b8504037ac53c90234d149ae : SUCCESS in 51m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6af99da9ad024db1897e2eeac634f9fa : SUCCESS in 1h 41m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5212be7725f24b82b2cc14ea5c141c70 : SUCCESS in 1h 42m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a9cfa9ea11c34207970dddc4c40e14dc : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a5af7a93f995af83ec98e71cbfdb44d87102eb8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd51027636f1c04a01261ab7056a4c353c74ed5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:44:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a011f187c0f74cd99a9019754040c87e : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f8555822ea04b03b82fb7f42176bcc9 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a6d412ec4b64a219ade0703eac36f19 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8055275c35c8496885626a1fb3118a99 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1187ccf931c34a81b5befa56ba6a8158 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4daad468a199476e821a2cc13d538722 : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e0123655e63046bb92d41a3c60ba99e3 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/327c75d8066146659cdb2ddedbadf18f : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b0f5fc9e48d45deb63c250827cac100 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92e1afafde0340ed8444743cd55ff4dd : SUCCESS in 6m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e452c1538f04daba85cf91a3e6722b5 : SUCCESS in 15m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/83b6c6cd3cc14dac801c996f8d040ce2 : SUCCESS in 27m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/24c0e97f8674434297f125b5d7fd7fef : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/202005abd04f4571b19f2e8595622f07 : SUCCESS in 12m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71e3b710713c4e81a6ff001901020685 : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/34668034a9aa462c9b8d47c5860ff612 : SUCCESS in 5m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/92091ec6b9844a64874db284a82a8162 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e67d407378a8452fb47bc16a86acbefb : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af2856f4e1fc498ebd1659aa5c5afe35 : SUCCESS in 13m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/29b15951f2f04b81b7021052ba8a1a2e : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b337eb3a6134f009eda3482f201d075 : SUCCESS in 47m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b44d501fe3ec45bdb9820760912f928c : SUCCESS in 1h 14m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00799d2e545c4d8f9c538c341dd15ebe : SUCCESS in 1h 23m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/c01a501b5ef8470db61937f69bde713d : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6b671091c2f00e303049c3519919160e1e66b920","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:45:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"007a287c99738cb9ac93d49849ea746a56be8703","revisions":{"007a287c99738cb9ac93d49849ea746a56be8703":{"kind":"REWORK","_number":1,"created":"2021-02-08 13:21:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/774446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/774446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/774446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/774446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/774446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/774446/1"}}},"commit":{"parents":[{"commit":"6c63edff2233e889a775320106a2a22b1466e8f1","subject":"Merge \"update constraint for tripleo-common to new release 12.4.3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c63edff2233e889a775320106a2a22b1466e8f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:21:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:21:28.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 3.1.2","message":"update constraint for oslo.serialization to new release 3.1.2\n\nmeta: version: 3.1.2\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I712e9ad4b4b998b80ca71fe53fea1a6cbf328e3c\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I37a4d48f621499b2de238cb9b22afb0528772d74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/007a287c99738cb9ac93d49849ea746a56be8703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/007a287c99738cb9ac93d49849ea746a56be8703"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
