)]}'
{"id":"openstack%2Frequirements~774447","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I99c0f51f4f4d7e006d9cd5ab580d4a34033d1c4a","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I99c0f51f4f4d7e006d9cd5ab580d4a34033d1c4a","subject":"update constraint for oslo.serialization to new release 4.0.2","status":"MERGED","created":"2021-02-08 13:24:04.000000000","updated":"2021-02-08 16:44:05.000000000","submitted":"2021-02-08 16:44:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774447-new-release","meta_rev_id":"210401c228b55d0a8e2e1eb84d63d6c243b01beb","_number":774447,"virtual_id_number":774447,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 16:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 13:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-08 13:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 13:26:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 13:25:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-08 13:26:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-08 13:52:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 14:48:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"090e791c41534be30639a06a72f441e261560163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 13:24:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30adcf3b85e501b47c9c93ca469c90ba79bdd087","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-08 13:25:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c4076eb82bc79d822ba7506dcff2cc773ba6f37","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 13:26:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1de503d8016cec87ff2664af453ade881f34b6ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 13:52:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c09f308dde50486b9336029820935959 : SUCCESS in 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1783dbbebe00f4f60f0302480e3b86ed1ccbab06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:48:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87f726181484bee99f2eb0522cebf27 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4cc2e000c174e4290e55e5acec9e330 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38caaaf657094d20b5f622086d6c8c73 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02a0d23594244170b8a574bcf1ec265b : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d45a773a11af4018ab7b6e91b7976fc6 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/531e25aa0331478eb08de5d585b8a76c : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b3429ed4782c4ae0a209120fdd85c650 : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f78356de7e354a26a11cb3acd936ae3c : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50b05400420841baa613c0f645cc4d9a : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e460416491274987a709c02223142c01 : SUCCESS in 7m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24603a20aecf44a0a8d36431be41011a : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e5206e9264a84a20a4c6545af2bfffb3 : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d06ac1f4c6824562bfcce881137b2d2c : SUCCESS in 12m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71aadeaad8ab4855b0c12039990e1d5a : SUCCESS in 29m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dcd95e5c23e0412c9f51b89b6928b764 : SUCCESS in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/87de0f3d5c2246f9bc2faf7d016d79fc : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0eb4c448e32499d8dc9fff28b6c3d11 : SUCCESS in 18m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bba84646d0f3413aba49c1ea1c4f1f0b : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6ff3c824ade4986b1eb1f2301c02218 : SUCCESS in 6m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b7550bef4574c878e1328a394cd7813 : SUCCESS in 9m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5389bc1eaac24c52b0b76b505538246e : SUCCESS in 50m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fff593402eca42189919ce72ca37f9ef : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"04a767f29c78585dd34394c48040565ec9d57938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:48:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6696f873e3ea3c49771107222bcd1de49d434d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:43:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2765db44e6f4b0e993d67b7ac2846d9 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23c5f72373f04b399dfd0fc531b60165 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dfd2cb7aee04e7cbefbb8525d039c70 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc6e62676d4141b5a5a291b420280040 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2459621b0d61481d92e4066347ce6998 : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e16a899be0b84ab38cb9c8330cfcaebc : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0bea8f5cec224db9bc2097bcb44fa3af : SUCCESS in 11m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/68c3c19deacc442185213c3e12911a4e : SUCCESS in 10m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/69bb39676bb342a59f9e034a2dc03de3 : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6cb1895cc504e2e8eadbc5d89385765 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d01ecd59aba940d399fb82af0656d645 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d8f02f1849d5406aa1bde13b3b71add3 : SUCCESS in 7m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/78c9eab14cbd4ce4b28d423491baa924 : SUCCESS in 14m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6791876cff244aacaed9e7d244d97eae : SUCCESS in 27m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8ff21b371841468f95636b298827206b : SUCCESS in 11m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aee660d3798442d39b88c33ec5c36fd3 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/711c9cace10747ebaaaf931b150d62be : SUCCESS in 19m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8357b51967f944e89af5173a12e11cad : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c33f62f6921742d5aafbb1210254fdd3 : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/346c6e4df12d46189d16bdc66a9e12dd : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83e30723e67e4dbe94eda04cd83a5b2f : SUCCESS in 52m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d17b94a430f24f04af1c25dd6ea16937 : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"210401c228b55d0a8e2e1eb84d63d6c243b01beb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:44:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5840117b74ac625d4eefaf6a4a73654e0777c06e","revisions":{"5840117b74ac625d4eefaf6a4a73654e0777c06e":{"kind":"REWORK","_number":1,"created":"2021-02-08 13:24:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/774447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/774447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/774447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/774447/1"}}},"commit":{"parents":[{"commit":"03a8e8e3fc90a5249371921d98edc5633440763e","subject":"Merge \"update constraint for sushy to new release 3.4.2\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03a8e8e3fc90a5249371921d98edc5633440763e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:24:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:24:02.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 4.0.2","message":"update constraint for oslo.serialization to new release 4.0.2\n\nmeta: version: 4.0.2\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I712e9ad4b4b998b80ca71fe53fea1a6cbf328e3c\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I99c0f51f4f4d7e006d9cd5ab580d4a34033d1c4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5840117b74ac625d4eefaf6a4a73654e0777c06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5840117b74ac625d4eefaf6a4a73654e0777c06e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
