)]}'
{"id":"openstack%2Frequirements~774450","triplet_id":"openstack%2Frequirements~master~If35561726af72c8cf15b493e8099d2fe92bf689f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If35561726af72c8cf15b493e8099d2fe92bf689f","subject":"update constraint for oslo.vmware to new release 3.8.0","status":"MERGED","created":"2021-02-08 13:27:47.000000000","updated":"2021-02-08 17:30:45.000000000","submitted":"2021-02-08 17:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774450-new-release","meta_rev_id":"dfca25ff4b75f2315ffdbfa8797ced992852c842","_number":774450,"virtual_id_number":774450,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 17:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-08 15:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-08 13:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-08 15:27:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 13:29:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-08 14:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 15:26:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 15:27:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b400bd53e7296ed97b63bb73e670e02489a4ee0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 13:27:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ede8446bc67e72055bb7c6e6edc4fc62071d0763","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-08 13:29:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d17cabe15e59a707a4a5b4afff1f2b520f742391","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:12:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/354b889b5984428496ef59ba36926c7f : FAILURE in 38m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9cd58195b5c30e2f4286acacf514cc39786aaa35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:26:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5aedac49f94440ba0ef7114359c457 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0286d84481bb407589f43acfaeb54884 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26878870ae4947ae9d13e7ce97f92086 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3d750f13a7344a483536598f009219d : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56c90f7259a84ec88422e9badf1c8c17 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82ea62816a8b44d7b1292572fbcc61b1 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28655931a768422886618d920cc07627 : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/253e44e6ab214d30a08521864f575b51 : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb5d84bed98a47ea9878f5ede172693a : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8287b588995247e0b45a9108096fd2ae : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/706a53043ba749ed83bf0aa852c715bf : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/286940b1db9d40a68cf02f22527fdf65 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cea6920466ed4fceb30d2feabd85a8f6 : SUCCESS in 6m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/430df6be18094910bcd6bf3c5986dfbd : SUCCESS in 6m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70c909dcdc2b4eeb821942924a84ccbb : SUCCESS in 12m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9e4e355ea35b466d9176effb45ed7160 : SUCCESS in 26m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/719400ce7089439e8ca3683e71fe1f73 : SUCCESS in 11m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56df01c250a64be2b4e2d0fde5145a7e : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c86ef95118046aeb9bd9a52f6af9db4 : SUCCESS in 19m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/adffdd13e5e943b79f40fba2e6a5b64d : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cac391901d6a4ea0ababb7b4716d5e8c : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ca3706fcb28b4a67b372267652ae7e0c : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fe1773f3ec349de80e4838db121ff4a : SUCCESS in 50m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/697867fdf3df44559923572e311ee869 : SUCCESS in 1h 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"95724ee11b7e078938ebefbc4734380b494afaf8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-08 15:27:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46958ed52a31d5600c1a0fca45b577b3f5af371c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:28:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b110f4fcd8e3ff4d2516cab9a1eea638c6753f6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:26:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92df2f79b8441e4a8242df888abd39d : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940cfc144d3b4150beed4c790470e73b : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/757a272a76404849abe575da3595ce1c : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12216f44b80e43b08afd26178d593a0e : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8d5d2040d764ff1be81191fcb44b670 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/852d1492a8d5490b82c6d051a27179be : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d877957f72cf48078e558f4edcfd7cbf : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f6bc6de3e0394e799d5d5ef8d4752c25 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/081252f3914d4732a419de0dac7ac561 : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4d7877305b6e4864a21e510ed8069b5d : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cbbda0f071364929805d81974e1eb9fa : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d2be22c4198469fb1acbc151a534e02 : SUCCESS in 7m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e11248014fbc46ae9adac2326954a7c7 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aa0dd556b6a44055b2019bcab68160ec : SUCCESS in 12m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bef3cc49d8b84ff1b7551cf0212864ed : SUCCESS in 27m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a1a8f8e798a04ee99e09159bbb74771d : SUCCESS in 11m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6919c24596f54cc6a15e34e674a5b3b8 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cf49e0ae2e244d591681899c4b7b186 : SUCCESS in 19m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/543d7d631995413fa3fa1eeee8f45a6a : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb0be6a9b14345cda7a0429f1beb3a49 : SUCCESS in 6m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a6520db3e72a480e8b0818b514903639 : SUCCESS in 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85e20dbd612d413289a7392fabbcafff : SUCCESS in 1h 01m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52ba9bdec0094de7b726fda38387f255 : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c629f5ef0b5123255f5dfd58da41bb6892bcc9bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:26:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfca25ff4b75f2315ffdbfa8797ced992852c842","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:30:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322c96d6ac78473883e7def280c2a763 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d6dbf69fc25c95c70f86c1056cd2c699a6ac925","revisions":{"5d6dbf69fc25c95c70f86c1056cd2c699a6ac925":{"kind":"REWORK","_number":1,"created":"2021-02-08 13:27:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/774450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/774450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/774450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/774450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/774450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/774450/1"}}},"commit":{"parents":[{"commit":"ed019e17dfc136092fe55ff41e2570f566e320e9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed019e17dfc136092fe55ff41e2570f566e320e9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:27:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:27:44.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 3.8.0","message":"update constraint for oslo.vmware to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I712e9ad4b4b998b80ca71fe53fea1a6cbf328e3c\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: If35561726af72c8cf15b493e8099d2fe92bf689f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d6dbf69fc25c95c70f86c1056cd2c699a6ac925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d6dbf69fc25c95c70f86c1056cd2c699a6ac925"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
