)]}'
{"id":"openstack%2Frequirements~774461","triplet_id":"openstack%2Frequirements~master~Id3b51761805e3af00c2b7e5f0e90459c3fccdd64","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id3b51761805e3af00c2b7e5f0e90459c3fccdd64","subject":"update constraint for oslo.metrics to new release 0.2.0","status":"MERGED","created":"2021-02-08 13:56:05.000000000","updated":"2021-02-08 17:54:29.000000000","submitted":"2021-02-08 17:52:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774461-new-release","meta_rev_id":"3fc2448315964c84b51bc3ffff13ddf1213f7795","_number":774461,"virtual_id_number":774461,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 17:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 14:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-08 13:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 14:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 13:57:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-08 14:35:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-08 14:45:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-08 15:37:53.000000000","updated_by":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"reviewer":{"_account_id":32523,"name":"Ching Kuo","display_name":"Gene Kuo","email":"igene@igene.tw","username":"genekuo"},"state":"CC"},{"updated":"2021-02-08 15:58:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59ec12d4c44e806eae58469f39ee442b988ab880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 13:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0574ee9abc2d04f1ddce3bf079a4fc2bb5c3877","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-08 13:57:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3504908d0ad2f8ec815a7c89ac326d1eb51e9c23","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 14:35:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbc0e90da5f6a123388e49d45d51696ae9d3dfd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:45:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c74ac05dd8f040a48913db77b9eb6745 : FAILURE in 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1f1529d6991383718579f6b0d60caf87fe87e06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:58:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5b0320b728411381f544ad5a79b2da : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ed0a4b867e41b6b8c66661012f0950 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dd805885f454d0da09f59b2c5b5d5d8 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03d425108c9a4ff0ac817267cd1a648c : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/696335ca96c142458f326498f825d0f1 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8ddfbcdd29a40ab8684a476bc0051b0 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0e25ab88f024d6da39ead5f1a7fdf63 : SUCCESS in 3m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ecfcc98c7c624b8aa799529b6b2f13b5 : SUCCESS in 7m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/11a8fae3d41f4b01bc45c3fdf08b2edc : SUCCESS in 8m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/173e06eaaad044f4923dfe9dc14fe7ed : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b32c37491dcc480885330c35bf2d76b0 : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c69d91beddac42b6998de39d4b3d6576 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5424d13e20449b78e98bc3b23dbde2f : SUCCESS in 7m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/064b4339dd0e4e5ea167463d87a77b5f : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19a5499a40bc4dcab9dbb3121cb30192 : SUCCESS in 17m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/529ccdff0b954ea0a8dfe515b8e14e45 : SUCCESS in 27m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a20c6f4c528b484aba8e475515ada497 : SUCCESS in 15m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e2c977f505e402ba15a18f3928d1b81 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb3bd037e47b4632b8eafaf318f15dcb : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0e74089ba8b44422a30025eeebe131ef : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92058320525a4d96b7d4d1f9c63d1a65 : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/41aba3b8185b4ffc9adf291401022eb8 : SUCCESS in 15m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7a8656004ca4eb9a4f3a5669e341843 : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/411acd5d674043ba994c84e90f6922e6 : SUCCESS in 1h 36m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ccfe211ad073864eef0de54884e9ef343dbc37a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:59:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"885fa1957b974c50f888736924925c2a788ec875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa5b49e7cd34ecc9a82d41f811fbca8 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd93a99fc3c486e9c1492f1e2801a7d : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70d2c2431aa5404091276a2692c78380 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06039f0e3af4493e8d393be9a70e21f0 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f918a9bfa8f145988a196e7e2d5859f7 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a1117db6e7444ec99564184e6e62131 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f6d671130f747d88041628b611d1964 : SUCCESS in 8m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0aa3d4d5107d47569f3147211e8e8532 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b0f30a73d694c0aafb6078265dc736a : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5e3203508f264bd49fc8ebddbab94e3f : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6910879316848f9ac3bb774f5ea2ded : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ce22582fe014e8587de65ae327bff37 : SUCCESS in 7m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/521a0fd4f76d40c3a499a79838ffbf5f : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d535190ee994dd599db1366024fb196 : SUCCESS in 13m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61d091186a8f4e74a0a071ee51642620 : SUCCESS in 41m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f458a3cdfe0e4ecaada155b4dfda184b : SUCCESS in 12m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08884c0062b64625bf7398b4a28a8aba : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2bc46f59827d4122880947438be5f8ea : SUCCESS in 23m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5878ec445bee47f4aa9bf65de520d652 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a73dfb6e78d04df987e41c849a5cf930 : SUCCESS in 7m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6be3adf959f54468bbf9fd807ec196a7 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28b08d0b0e3344d7b057e226a51770c2 : SUCCESS in 58m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f9537fa19846fc914b5604934e96db : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b50f1636e99cda1cc7b0d86cb548f451af6fe544","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:52:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3fc2448315964c84b51bc3ffff13ddf1213f7795","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:54:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97da168381e04346b8d521013eabbac5 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1379c59249c9940c9582b752463df222ec76dfef","revisions":{"1379c59249c9940c9582b752463df222ec76dfef":{"kind":"REWORK","_number":1,"created":"2021-02-08 13:56:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/774461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/774461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/774461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/774461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/774461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/774461/1"}}},"commit":{"parents":[{"commit":"ed019e17dfc136092fe55ff41e2570f566e320e9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed019e17dfc136092fe55ff41e2570f566e320e9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:55:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:55:57.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.2.0","message":"update constraint for oslo.metrics to new release 0.2.0\n\nmeta: version: 0.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ching Kuo \u003cching.kuo@linecorp.com\u003e\nmeta: release:Commit: Ching Kuo \u003cching.kuo@linecorp.com\u003e\nmeta: release:Change-Id: I68dbea306109ed25124480348a00fe3830066868\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Id3b51761805e3af00c2b7e5f0e90459c3fccdd64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1379c59249c9940c9582b752463df222ec76dfef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1379c59249c9940c9582b752463df222ec76dfef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
