)]}'
{"id":"openstack%2Frequirements~774462","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iaac32d3edc938e568b454bc5359da3a6bce234d1","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Iaac32d3edc938e568b454bc5359da3a6bce234d1","subject":"update constraint for os-refresh-config to new release 10.4.1","status":"MERGED","created":"2021-02-08 13:56:58.000000000","updated":"2021-02-08 18:55:46.000000000","submitted":"2021-02-08 18:55:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774462-new-release","meta_rev_id":"27b516fe55d697869cc4fb508ab2a0fc8460d6bb","_number":774462,"virtual_id_number":774462,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 18:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 14:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-08 13:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 14:34:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 13:58:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-08 14:34:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-08 16:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f89a8d763bba23fc3d43f0e45fa00b89d1e73392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 13:56:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3275f69f74acd4e5b87c47c5b5c09f745c0f993","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-08 13:58:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c66096e588a0cd78a600405257639b38009b062","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 14:34:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ac0c3cd60d5636bfef88e539670ec46af5c61b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:07:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5ffd64672c9428fa05644693c6fe17d : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1609dcc5216244c7903d3bd388763e4e : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6826952831542a58a21e6c9ed9ce94e : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40f25471367c4e219399258bdbef14ee : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa3668ea78fd492b8feb3f83c4d28012 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c64507826c64f2caed166a42afd5a82 : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5baee67dc1a64084befd8be306f52979 : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92c53ef6b4bf4121ab17470475b0f84c : SUCCESS in 8m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c0a64ee873f94eddab0f3faedc1c22ce : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18174a22891146f1bd427f95607025cf : SUCCESS in 10m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/613c43b7d6084506ac557d434283615a : SUCCESS in 5m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2e0e674adff432fb0025c1aad539e76 : SUCCESS in 5m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/685d3caff1c14e99bcc247443c99e577 : SUCCESS in 9m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b978c4aaf78945aa9433e3d31bd39ca8 : SUCCESS in 7m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d8b921fa7954a1fa081897a8bb9b5cd : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e25ae924a50e47bb85f9214fbdbc5412 : SUCCESS in 15m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74e9fb46dfb84bbcb7b5fdab8e7323d4 : SUCCESS in 12m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b99a17ea652b40f0a570b683a184fd86 : SUCCESS in 22m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/27591d51d9864790b905af9cfccca866 : SUCCESS in 21m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/258672e7761242fc8d5f00de97d6071e : SUCCESS in 12m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ec7eeb11d5f4024835fdafd390f7d98 : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/05add4d7503742ff834fa679249c4353 : SUCCESS in 12m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2900b67e3224b2fa949ad0520796bd8 : SUCCESS in 25m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a91f6e3a9ba4458ad2b75249f6f7bb8 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/afb485b170e743d583cf115f16d9424c : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ab921426f7445edadcc8e9cdcf278ed : SUCCESS in 9m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f178a665f4a54d47b586289927f4b2ca : SUCCESS in 1h 23m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a832368481524d50a7e22aacedf0c3c2 : SUCCESS in 1h 45m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1746eb24eced4737946d84b6396604a3 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d93ac0ef83eb5d2676ac0d208a15734d188cc9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:09:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd7ce2435e560e320aa9094590a70437ec995791","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 18:55:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d19ba06a59408a92eb31a49d238287 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e52db1b03fc40e89c33dc198d561c06 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b00fedb70a314d7cbaf1cd7703263eda : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f35d098c0a044491b31ba6a1dbef34e0 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec1b23ac98364217897bc4eeb0116ab4 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c56604d67c341fcba14f0de5cd63ec5 : SUCCESS in 4m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6e5d08b9b6e24fee98c5fdce4d8a101f : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc09f6b2a89c4f0f8dfb69368bd9aff9 : SUCCESS in 6m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4ecf71690189482a9bdfaa5eda6fd00a : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a64bc4361544ff7872a5fd84cd42548 : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1ca48717d1c5494e8d8ca8bf71c9b1d2 : SUCCESS in 4m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/28c7898a140c41bebb3c41c0565d48df : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0ce3fd40a7b34aa29db4d37f209da4eb : SUCCESS in 9m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2a46f1acd7a4ab383604ae6e8f505e6 : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b194fce5bd834bdea8ac991a632ef176 : SUCCESS in 6m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/87fc6dd08b0a44fa9afb45991410ee2c : SUCCESS in 13m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4e9ecbe53c344cdb8c69ecbc866b5ad : SUCCESS in 19m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bb8a041f54ce4808b04387484f0d6081 : SUCCESS in 26m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/72e082f16673487fb05f53a28594b1be : SUCCESS in 25m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a1dce33f2ca640bdb2122671b9e6d5ec : SUCCESS in 13m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e92b4f6003646f9aa776a040972f71d : SUCCESS in 16m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fd25fa0db75e4eadb921cfe11a9a990e : SUCCESS in 18m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d406c1b470e43819f2f60ec90e54914 : SUCCESS in 24m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25121fa68a074f1998270129bc93dac5 : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cfaf6d478cf4430a8aa05f525ce2bc71 : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a2440cdf2f74bbe949e9885b63d799e : SUCCESS in 12m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e2989173b6d42bb828cac97e8c997f6 : SUCCESS in 1h 03m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed8e1248712a4e119522518eeecb58e1 : SUCCESS in 1h 35m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9cb72b17bb7442cb9bd2b826d1426aa2 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":1},{"id":"27b516fe55d697869cc4fb508ab2a0fc8460d6bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 18:55:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef0b8aca68b279bf5567fe5917bdbf036b514079","revisions":{"ef0b8aca68b279bf5567fe5917bdbf036b514079":{"kind":"REWORK","_number":1,"created":"2021-02-08 13:56:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/774462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/774462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/774462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/774462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/774462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/774462/1"}}},"commit":{"parents":[{"commit":"49a467246a83acdfd4f5c163dcbb28c75044547e","subject":"Merge \"update constraint for ceilometer to new release 13.1.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a467246a83acdfd4f5c163dcbb28c75044547e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:56:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 13:56:53.000000000","tz":0},"subject":"update constraint for os-refresh-config to new release 10.4.1","message":"update constraint for os-refresh-config to new release 10.4.1\n\nmeta: version: 10.4.1\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1348f2cb56f7615a9be9a933f74bc487dffb909d\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Iaac32d3edc938e568b454bc5359da3a6bce234d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef0b8aca68b279bf5567fe5917bdbf036b514079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef0b8aca68b279bf5567fe5917bdbf036b514079"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
