)]}'
{"id":"openstack%2Frequirements~774464","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I30658f0da3fa8bc312088eceac1eca574404cf90","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I30658f0da3fa8bc312088eceac1eca574404cf90","subject":"update constraint for os-net-config to new release 11.4.0","status":"MERGED","created":"2021-02-08 14:00:43.000000000","updated":"2021-02-09 04:55:50.000000000","submitted":"2021-02-09 04:55:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774464-new-release","meta_rev_id":"8d61fcd5a795ce424302e3f9f1db03398d9823db","_number":774464,"virtual_id_number":774464,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 04:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 14:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-08 22:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-08 14:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-08 14:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-08 22:50:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 14:03:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-08 14:33:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-08 16:13:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 22:50:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bee7a606bc2740ea6f377787da6fa4e73ee15143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 14:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c073fd76cdf6e784e6af49d1785ec02859ecdb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-08 14:03:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60f0bc971036054c16b06d0567d673663f12359e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 14:33:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ff4f535bc33dc1fcabe2b1f8e05923c06de7459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:13:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b954c9eff7a4c209f2659255aa3bd30 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2939a3dc42354e1c8b9b3ef9a16b96df : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b8ac02ee7e44375adbbda4d14deb3e5 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff457bfddac744b6a7a4136c7426fcdf : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7686e1b63b2f435fb10793aee09b2f8d : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3dd4d529b2f14eceb35550f8527a6d69 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b17d76f1bb164fb0bdd1e7107fb6b0ec : SUCCESS in 7m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca8823d275fe495199630d990c1a25ab : SUCCESS in 6m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/14004387632d40f1a7381ee6ea0290ac : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aefe8adb3e234c73861321bf4dd6b263 : SUCCESS in 10m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/df01ee4988124de2b4d118a4b6bb4a89 : SUCCESS in 4m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a75f646391f44fd80d25c9b09e2cc18 : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3a0a9278dc62416b9ff363bedfee8c52 : SUCCESS in 10m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9f1d6591ace84c5f81c9edf43f8c9273 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/707a4afab8364d8bb8611a98ea9d9eab : SUCCESS in 6m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6ea8e4c076a54eeb902f8a5fb7ee5c53 : SUCCESS in 11m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9ef44915aa41475f8b313db3cc9d812f : SUCCESS in 18m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/95e382cac3864dbe901fbefa2648ac6d : SUCCESS in 28m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/84df51c48bc34d63bb681e17c1093e26 : SUCCESS in 27m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/32bde14c7f514d538aed0354a79730cd : SUCCESS in 16m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76983c5b4e554df39178848d2c6090c0 : SUCCESS in 17m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f094f3226c3140c5ac3ab7141d9052ec : SUCCESS in 10m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87c7b3c99e3f40b9ab17c4bed478091c : SUCCESS in 15m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9067803110804f2f8b564dfa06e51860 : SUCCESS in 4m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f63603dcd8be435bbaca21fd25d2a1dc : SUCCESS in 8m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cf494ef37ce84d64a50beda909098aa5 : SUCCESS in 9m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d7f2857d3944a589e466366de27f7b5 : SUCCESS in 1h 03m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed5cb42a36ea4112b31cc9336bc7bb5a : SUCCESS in 1h 46m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6e91d88f63bc40dc99d969f53d85aed2 : SUCCESS in 1h 05m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e7556aa8e4e4f7e65576216e8d2e3551d92dbdc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:15:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26c9406b27947631539a8c982d7621129abdc339","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 18:55:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73faaece02c34e799d5414dfb93850b2 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce2e31f8d74c4e6ebd9672d40ad8c001 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c7b4bd376d7400e9633772b704a351a : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bc1b2fab6264fd0a57dd20fbef69a2d : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ca2b6957b134597a5a69a6b61bcc3bd : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f060d61749bc456abcb2aa2a75d815f1 : SUCCESS in 4m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/73dccc87a70c42bbabdb95a1db759d6f : SUCCESS in 8m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c1e5524f5f0546c7ae01119105b2d749 : SUCCESS in 10m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/08aa3ceeef594ef7a0cd3fe36c62d71c : SUCCESS in 11m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ac82ab6ab87940cd8f0f35939413db31 : SUCCESS in 10m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/04c9a85f44e545e8b431ac6d30389a06 : SUCCESS in 8m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/82c62b9dd0344f9aafe94e4a990b783d : SUCCESS in 7m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dcac4a2506d844f9b355ac30e3ff32a4 : SUCCESS in 9m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53d2d89f8bef46a2bf61be81f7add5b3 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b1a5e571c9c45c5bbea636421dce766 : SUCCESS in 8m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b543e0a1274d491fbf668f76c672a9ff : SUCCESS in 19m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb4327515069446fb5fb0ccd3b9ae2e3 : TIMED_OUT in 1h 01m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e35ebcb96fe44cfb9e311ab7a22cbdac : SUCCESS in 17m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c4ff93d12d964986af78d1d8f0120804 : SUCCESS in 36m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9625b58e0bec4f30b81b732c81df393d : SUCCESS in 12m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6beaead6005b482ca227c33b988a3c00 : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a4452e55a4c743039d64de5f0ea45981 : SUCCESS in 13m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95aa7f8104b440d1ba5771be8b70dace : SUCCESS in 22m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6fc61e52a3144f3abd0efb75f1fb6503 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/74e5c4f787cc4c35bfa6ff96f0aa9a34 : SUCCESS in 9m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f58768dfd6744af0a24959791600da09 : SUCCESS in 9m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/953ddb22156e4fd8845c4cd0f2e41f79 : SUCCESS in 1h 05m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a674a2a06f742aba6fe6f8ee0b57fbf : SUCCESS in 1h 26m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b8490848b58748e68ef45039b32dfe29 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":1},{"id":"91f9f9a13507daa64d0fdde2b104151e62f45c86","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-08 22:50:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9011430f7fa94d98783115ba7e4dccd8bc7bbc2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:08:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3825e3290914358926d4a0a82725a25 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de0a9e3dc4f94cfc8c3601bf22e80b44 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/041d96a7a4ab45a28b0f7becf89a3630 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db2859be0a1d4b3281b6716e687a6d0e : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59d495d6629b410ba176ea5c699abd99 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d6cc590ef5b4e94a5430fac2db3a7e2 : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/db8bf1fe333743bf9db49465c60b3bea : SUCCESS in 7m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e642215a5d3a41bead400078ff006568 : SUCCESS in 8m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ae55133359b2448fa66de8b222ae6249 : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b4479822faa4dc28e1f2fc268001c4f : SUCCESS in 8m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ccfd7bfc7b5e4cf79111d1006c57411d : SUCCESS in 4m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ba799953485d4b4489a29969a76ba861 : SUCCESS in 4m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/11b266c7780a4361b14443aa0394bf5c : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5e0cd0c2b75487f9c89310f73540e1e : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bb1f846d36444888ca414b35c5104c2 : SUCCESS in 6m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ad9e0328fbd14a2498891cbc44eb5fbb : SUCCESS in 14m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7a2fad6409441378eec3ebf80fdb728 : TIMED_OUT in 1h 01m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e78130d237504de59f91ba79942ff962 : SUCCESS in 22m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b8c98870b2ca4be6a12926326f0bdb31 : SUCCESS in 27m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8559418e99dc46498edc8e3e4d0b99dd : SUCCESS in 12m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5815097b79b640f7a1200481544fa102 : SUCCESS in 11m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c85a95ffabc9479eb7a35dca0bf3fa31 : SUCCESS in 10m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88108db65e6f4123972c1d05cbc718d7 : SUCCESS in 19m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/541c13aa78564a60a8957241b9ee1b01 : SUCCESS in 6m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1d05ccc0676642d2b007a2e9203f685e : SUCCESS in 9m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/288d0e95468048448479d2b1171e4dc2 : SUCCESS in 7m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3120f6ad14914077a6e82d85e8b535fa : SUCCESS in 58m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/37546eea0b89422594a11c7a40ff646f : SUCCESS in 1h 15m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/59ed30e0ade74bcea1b91d3381eb226e : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a36867256b1da87e39307afa91221f51360d16a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-09 02:13:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41f004c253d3a6242fd70f814cc63c8cfb24a735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 03:41:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47e033de6c54d68914291dd0007f7c2 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15b3c2614a664bc0ada3ece906dcf60c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba5b0207e3b241f4a57eee5859c65f84 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38ec4aef59b64684a43292f5fbb42c21 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae35991e83374abebfe6aa52078d2284 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7040cb99707247808390f2ca622abf07 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/27a4e2d30f6342ca87354bf0ab4069f1 : SUCCESS in 8m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c10a8c29f0d4d6aa5ca9898562aef54 : SUCCESS in 9m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bd10aeeb89da4ca692dd8d8f7e9b0f0f : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3de3873c85944497a296227a1a628127 : SUCCESS in 8m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d22dca85fda142d4a842c4b54eea9b2a : SUCCESS in 5m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c1e885a16ac4bc6bbe229ec661fa858 : SUCCESS in 5m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4749d1d6750a4ee899a63c381f63771c : SUCCESS in 11m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/646ddebabe4a4d299dff4dbf91965880 : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/872d7d7238ff4ad1b7cf18245c3c4660 : SUCCESS in 8m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fc65e3711d424685959219a3d6780075 : SUCCESS in 11m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ff1c8b6f3be4b3cab35c9d65d884f58 : SUCCESS in 21m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/95c27e2bff6e46389b57f0a8478c1917 : SUCCESS in 20m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ac372dbba51d41e7ba2717ef6c837cdd : SUCCESS in 19m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0eef99a7c8094fe0969952d1496356d1 : SUCCESS in 18m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c126a4e04e8d40f694982478786db043 : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/14075e70d6334aa5970fccb2eac557d9 : SUCCESS in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3007ca10dcee43d397560e168afb3592 : SUCCESS in 18m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c515644e00a4466a2083df116840297 : SUCCESS in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a7f71b522ddc4d92a5483bab49b0f428 : SUCCESS in 12m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/402d34c654e24486a489a3de7149cf26 : SUCCESS in 8m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/508148ddf6c441c3b40a2259a5fb0e3b : SUCCESS in 59m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/407d7eca3dc34b198e04afe4d7d9fd8e : SUCCESS in 1h 24m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/40bb4f20027c412cb66e1cebf5237883 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1a9119237617ec3b7999699439b3b10add11dd76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 03:41:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ade7df6c6c50cdf6400d2e1b7b7301572384437a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 04:55:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709b045117ed497b81f7ce79ae25bfc9 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac397935dcac4540b30dd8aec40acd50 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97fc2cd644da458584b1a514746951d9 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f40cb3bac9984a3e87db1bd92bbdc587 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78db2b3c109f4e788fc43f0505060e5e : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fe91c0be68f43d4a32f96850abb5daa : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c729242f94204244bfdf46f4cfbf9803 : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/922a3ba9e2a6488a8c6a91f16acace3b : SUCCESS in 6m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c99302643f734f658f33a0bb2d20fa8b : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dd301c25e0aa4d34a0a216f5f9442a62 : SUCCESS in 8m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7b1f6f524fc847fe9853274366fc63d0 : SUCCESS in 6m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dae5daa42e6b4b1b9be9b8e0202855f5 : SUCCESS in 5m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/34e24b6267164d85b12525ff976b82af : SUCCESS in 9m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/87fbed63553145a891cda08cbeb7332d : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88b8a90908c14e72aaa08940ea780b99 : SUCCESS in 11m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/af7255207b83482bba8a5d4840ba5bd0 : SUCCESS in 15m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cfd9dbc6baa84ddebe67ed841d9745fd : SUCCESS in 18m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9e6736dc39884acd9085c3fb2efc0d63 : SUCCESS in 23m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/28065daa3c43450a894d8d3836ac9aed : SUCCESS in 25m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4011137c7fad422db2e1d55d2a2166ec : SUCCESS in 14m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/005622ca1f334b199666f34e8a99d9a6 : SUCCESS in 17m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4f7a70e82aa54bdfba1cc449b4b4ac34 : SUCCESS in 15m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6237269bbd9b43919c33ebb21a20cb1b : SUCCESS in 23m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae065335c59b473e84c88c84ec2f0343 : SUCCESS in 4m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cca8fad1c84f469383845580b84da179 : SUCCESS in 12m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d6273079732947c28b8f42a9c6860c72 : SUCCESS in 8m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb2c03510f854797954e6ee0ba3b756e : SUCCESS in 56m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a43e549dd4f4153a5f5654806637935 : SUCCESS in 1h 12m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e30c277658446fd933c2a8c28fa6b65 : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8d61fcd5a795ce424302e3f9f1db03398d9823db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 04:55:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75757b1324471f7fb9c73d64e463cc90a1759e0b","revisions":{"75757b1324471f7fb9c73d64e463cc90a1759e0b":{"kind":"REWORK","_number":1,"created":"2021-02-08 14:00:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/774464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/774464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/774464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/774464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/774464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/774464/1"}}},"commit":{"parents":[{"commit":"49a467246a83acdfd4f5c163dcbb28c75044547e","subject":"Merge \"update constraint for ceilometer to new release 13.1.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a467246a83acdfd4f5c163dcbb28c75044547e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 14:00:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 14:00:35.000000000","tz":0},"subject":"update constraint for os-net-config to new release 11.4.0","message":"update constraint for os-net-config to new release 11.4.0\n\nmeta: version: 11.4.0\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1348f2cb56f7615a9be9a933f74bc487dffb909d\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I30658f0da3fa8bc312088eceac1eca574404cf90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75757b1324471f7fb9c73d64e463cc90a1759e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75757b1324471f7fb9c73d64e463cc90a1759e0b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
