)]}'
{"id":"openstack%2Frequirements~774465","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I4bf84bc29abba99e3e8ac8d68e90bd70bcbeed94","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I4bf84bc29abba99e3e8ac8d68e90bd70bcbeed94","subject":"update constraint for tripleo-common to new release 11.5.0","status":"MERGED","created":"2021-02-08 14:05:25.000000000","updated":"2021-02-08 20:58:57.000000000","submitted":"2021-02-08 20:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774465-new-release","meta_rev_id":"98906ae2b760a13a36f6121e1e3c3fac7cdea386","_number":774465,"virtual_id_number":774465,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 20:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-08 14:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-08 16:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-08 16:23:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 14:35:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-08 16:05:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 16:23:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bb0aa889fae6faa16b933f5f085f6e8c81c632cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-08 14:05:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c4de498a22bd0f941a652db8c2832cc4af11fb3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-08 14:35:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85e53ce9c2233f9679e378b013281c13d82f742e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:05:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96bef7f1a99b43ae8c005946190e378f : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa2a28ccd0f344ee9c4f66d39e7b05e3 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeb16f8cc70141c28cd26156c3648c4d : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/731b22464cd24b45a2fb0744c4315bc1 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da79bd1d1b9f4d76ac53881865d19028 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44a04255bc404fa1a7d0293b9c8d0c48 : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6d14bdc54e10447081562d156f5c2dc5 : SUCCESS in 6m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fdb3550181754099b4c4593f53858bae : SUCCESS in 8m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/778ffb7895d64424b1561fe3696a2fd1 : SUCCESS in 8m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3e4b388f89a4dd888e5215c37fb9375 : SUCCESS in 9m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ebcd8e6241ee4a048d423e11f24f6eda : SUCCESS in 4m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0323fdcd1e3d433581cc8dc3acd50184 : SUCCESS in 4m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/39bac1d3ca084426b3952b49fc9e8d08 : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f900d0a9e7b4fafbc520c8f63a4f092 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5abd78b7a494b1a93bd8772ed2846f8 : SUCCESS in 6m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/11e6a58cbadd4d7b8da5cd5b60122bc3 : SUCCESS in 12m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/56b28a0d2b5946b59f137d5ef922c072 : SUCCESS in 15m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6ad7340a5891440ab6026aabf116a155 : SUCCESS in 26m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8d5d70177f894ccda44dfa1d0ba36981 : SUCCESS in 25m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/78dfbbb5aa8843dabd1b3ae0db076ffd : SUCCESS in 13m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/132da9489cc94ddabfa4c6eeb5f5b798 : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/329abb19c8814d928986f436d37e712c : SUCCESS in 12m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e3573abdbbc4ff1a7ef9463fc081ce2 : SUCCESS in 20m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eebee9a875744254aed5a69b47c24add : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2df0529e0f0242d9b0049c0391bc4290 : SUCCESS in 9m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de8a525c6c6a411abc282b5576ba3066 : SUCCESS in 10m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d595176b32a24e319d194eadb3fea997 : SUCCESS in 1h 07m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/467b2fc700ba4c15a0892fe7401bb0b5 : SUCCESS in 1h 38m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/18b2a5a1c97541ca9791b13c84e47733 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4933bc6ec979f9fc6ba37b3f5b37bcc947e26c27","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-08 16:23:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d152b44b6ebc6d0d7a493126f48abeccaab9d68a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 16:24:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db705a30e6ae5098427eacd5e80de64a5cbd05bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:58:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/929a1f882aaa42669aa686085576ebaf : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06941bd92ee6459293bcc3d9038940f4 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b8e0bb515174434a8a324c6f37dced1 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/446916e458e24c7092829b1ede72657b : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c17f389c44d49b387a7910d454cdc2e : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/979f5b24d71840029db3524f5918a86a : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/73e42e5a10c44dc2b9fe5cc80404640a : SUCCESS in 9m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c39d9dcba95461bb592b922eb7ae329 : SUCCESS in 7m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/827fd1de8c944355b8653b3b68e1627b : SUCCESS in 11m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd8b6291d8e74a1bb0de67b37d8d78b5 : SUCCESS in 8m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/97d63bdc2bb840daa0991d38b16e8c9a : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6fb376479dd5445dbad71236fc0e80b9 : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5ca04e8205624ba4a2ed2a3ca120689d : SUCCESS in 11m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c32736c0c41142c5a2970f3e26dfb9a9 : SUCCESS in 7m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc66aa743b1747e0b6d967f3f7d315c6 : SUCCESS in 6m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d54ced960e6c4fb9b339afed019bae63 : SUCCESS in 12m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5aed0c5ce2f345a8b1cfd671cdadf353 : SUCCESS in 12m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/93209169238b4872a23431c2114ad43f : SUCCESS in 25m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4342ec9574c04002aecfea3b84fd0228 : SUCCESS in 14m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/085b545f6aff440598180b7e3615618f : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a4d082db17e44bf08110df088731f4c4 : SUCCESS in 14m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a1fb48a018104a38b194413ce3a72fd9 : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8ea397339884d02a5e586564cdf6125 : SUCCESS in 20m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b566226d0c7e4cf8bec349753ee14cd1 : SUCCESS in 4m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4f5ae009355f43fbba6c7493201cb0db : SUCCESS in 14m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85976e12ad914f8ea42b5c93a4d614e5 : SUCCESS in 9m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40f6f42bb8ec484d8aa7e2743d15dde2 : SUCCESS in 1h 05m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c68dc320a8a4167a671f08c57c98cd0 : SUCCESS in 1h 28m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f0416de0645b4c4b8950f4b76bb8de2b : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"98906ae2b760a13a36f6121e1e3c3fac7cdea386","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 20:58:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a9489aa828ef8a0eb9586367f9694eade30c634","revisions":{"1a9489aa828ef8a0eb9586367f9694eade30c634":{"kind":"REWORK","_number":1,"created":"2021-02-08 14:05:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/774465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/774465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/774465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/774465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/774465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/774465/1"}}},"commit":{"parents":[{"commit":"49a467246a83acdfd4f5c163dcbb28c75044547e","subject":"Merge \"update constraint for ceilometer to new release 13.1.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a467246a83acdfd4f5c163dcbb28c75044547e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 14:05:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-08 14:05:22.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 11.5.0","message":"update constraint for tripleo-common to new release 11.5.0\n\nmeta: version: 11.5.0\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1348f2cb56f7615a9be9a933f74bc487dffb909d\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I4bf84bc29abba99e3e8ac8d68e90bd70bcbeed94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a9489aa828ef8a0eb9586367f9694eade30c634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a9489aa828ef8a0eb9586367f9694eade30c634"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
