)]}'
{"id":"openstack%2Frequirements~774577","triplet_id":"openstack%2Frequirements~master~I1dcb7d7c25ad61500de0e0c278e366976d7c4ed8","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I1dcb7d7c25ad61500de0e0c278e366976d7c4ed8","subject":"Bump cryptography uc to 3.4.2","status":"ABANDONED","created":"2021-02-09 03:31:26.000000000","updated":"2021-02-09 07:34:52.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"29a050641617abc7397d19bfa0abaa2fc09e1156","_number":774577,"virtual_id_number":774577,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-09 06:28:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2021-02-09 05:26:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-02-09 05:50:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 04:02:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-09 04:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 05:26:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-02-09 05:36:31.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"CC"},{"updated":"2021-02-09 05:50:53.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-02-09 07:11:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"e036fc4548486392018c0814519ec668e8f55a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-09 03:31:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"100505d056ed3dd3fadbb1664c4090aac3b96b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 04:02:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4350c6aaa00749c186bb18b0f478c0ae : FAILURE in 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6f3f5d062f9637c7e55a92302817d55c6a13b969","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 04:49:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc9df490298474d8644565e55d1fd19 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d1051528c60403c93e9e925ee1a1682 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ab1a2cb826944d6b026db447b082f61 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ae5d8278fa24ddab554c3cb6aca551f : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e4ada178d874f4b9890c43844b1c9c5 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb42cece10be47df95f9a1dd3368f99b : FAILURE in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17db344673784f458192e0dc87ddfdb8 : SUCCESS in 3m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3058dc96bb7a4fa79263aee450cdde2e : SUCCESS in 8m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fd0ac35916f47218838d0db4b912a9a : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a4746e5c69740d98536b6643cbb845b : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc037d694a1540cfa0e2118012874ace : SUCCESS in 4m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a5e960a37f8544d3bb1223f6bfc39e93 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c404094ed5f64166b7f82f29b2b65635 : SUCCESS in 6m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bdba67d03f8b44958525150c0c05f910 : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e3ddf77a81e4981a1a2b6206e085472 : SUCCESS in 13m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7c775334f7c14655a24e9ad44f527ab3 : SUCCESS in 16m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7dffb9f02214d11a13dcc8e1d4dc47d : SUCCESS in 11m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f20d25a3fac478983bd0d2b9c16e114 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1806aea9ef44821887e8dcb0a8f618a : SUCCESS in 16m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/229230d6fc7440e69717eb6b549b8036 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd8c0eeeaa1e4baba36351aa26d9a18b : SUCCESS in 5m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cdd50f1925b4686a756a9100fbdf438 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf39afc83a044bac9e71f2ee7377c14b : SUCCESS in 46m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/416a5d906048475481afdf75f8181dd9 : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"428b2dc863932419f062c60a3cddbd5b6f0f527a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-09 04:53:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a929ffd2f3d4da69616a0e3b5c9818dc747660c9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-02-09 05:26:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"05ec6625f75813533c371bba0f1930cdc2677b45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 05:28:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2618be278d8941c39a9ce039bdbd06bf : FAILURE in 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"fa7c3c20e754ac67340f68d265e6739bfd70c979","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-02-09 05:36:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b109f93514e6a21c6fa1cdb9db942c6c5b18ad2","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-09 05:48:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05aee846d31927c3b5d2043d6ffac042fa7a928f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-02-09 05:50:53.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"6adbf7c9b3bcf872d56e27e0f291660be3bee36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 06:28:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebceb133e9445b69ea98d198750ece3 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0ca198f9a447b8b546408febea9007 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9847d53d28940ecbc5f3927a6e82291 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be07f7da46a4f1cb89ada6844523beb : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c25a0208137e49f39c3dfdf1ff5081db : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93f0f78ab1324d07b37a2bb34bf2063b : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9be6e66b7564e768b8025225b2864c1 : SUCCESS in 3m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9597aa27c5154becaf94cfe0cf99bc38 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/18ae9bb93eda49acaf076297d6eb1a76 : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c7e16685b434fdcb0f696bc5fe70a69 : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e47d275673f04edda352c09a977a49a4 : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d537d2d3e8ab40bfa3f9a19ed46f386a : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1a11749f0e94bbfabad333aeca31c8a : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/826b01d04df3472c8d810b6689f611f4 : SUCCESS in 6m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b6f6fe934e9f4588a2043fd005a43ca5 : SUCCESS in 13m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6e1c7773d17a43d096ffb424d9adb943 : SUCCESS in 24m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a36ec5394ba4077821f5bf8734d4263 : SUCCESS in 12m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e069dd486ca4d11bf0d2fda01d0c020 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/384233cfff614271af55e5b322e36b57 : SUCCESS in 13m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2f04b3307080483eb332b2e74f50b74c : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91456e2f9265410b94aecf3ad130bda9 : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b136ea11aaa4a8ba40dd5fa76c4c5a1 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0863e2732b5b457a8c6c3bf2d0668395 : SUCCESS in 54m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b8b5611f60048529fdfd95718c2ab45 : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":2},{"id":"448c6750dac2587ee241985b41a779893a1db995","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-09 07:11:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9216aeb7ec58c514a086f1e8daa33cf3f2f792d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-09 07:18:23.000000000","message":"Abandoned\n\n\u003e 3.4.3 is part of the bot update that is hopefully going to be merged in a couple of hours if gate cooperates\n\nThanks. That\u0027s good news. I\u0027ll abandon this.","accounts_in_message":[],"_revision_number":2},{"id":"29a050641617abc7397d19bfa0abaa2fc09e1156","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-09 07:34:52.000000000","message":"Patch Set 2:\n\n\u003e 3.4.3 is part of the bot update that is hopefully going to be merged in a couple of hours if gate cooperates\n\nWe also need to blacklist 3.4.1 I guess. Though our issue would be resolved by the bot update.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"239e47ff888be5d678ce3d9424b681eaf7a506a6","revisions":{"afdaa98b18ec49cf9f2befe2e756395adfbd6110":{"kind":"REWORK","_number":1,"created":"2021-02-09 03:31:26.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/77/774577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/774577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/774577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/774577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/774577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/774577/1"}}},"commit":{"parents":[{"commit":"3126e526ca932a1c66e2a9db6b392c217b7153f8","subject":"Merge \"update constraint for oslo.metrics to new release 0.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3126e526ca932a1c66e2a9db6b392c217b7153f8"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-02-09 03:28:12.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-02-09 03:29:59.000000000","tz":330},"subject":"Bump cryptography uc to 3.4.2","message":"Bump cryptography uc to 3.4.2\n\nCryptography 3.4.1 seems broken and has been fixed\nwith 3.4.2.\n\nhttps://github.com/pyca/cryptography/commit/843ada65e816a17e1b3d90b12ab6403c8ff96654\n\nChange-Id: I1dcb7d7c25ad61500de0e0c278e366976d7c4ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afdaa98b18ec49cf9f2befe2e756395adfbd6110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afdaa98b18ec49cf9f2befe2e756395adfbd6110"}]},"branch":"refs/heads/master"},"239e47ff888be5d678ce3d9424b681eaf7a506a6":{"kind":"REWORK","_number":2,"created":"2021-02-09 04:53:57.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/77/774577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/774577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/774577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/774577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/774577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/774577/2"}}},"commit":{"parents":[{"commit":"3126e526ca932a1c66e2a9db6b392c217b7153f8","subject":"Merge \"update constraint for oslo.metrics to new release 0.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3126e526ca932a1c66e2a9db6b392c217b7153f8"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-02-09 03:28:12.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-02-09 04:53:40.000000000","tz":330},"subject":"Bump cryptography uc to 3.4.2","message":"Bump cryptography uc to 3.4.2\n\nCryptography 3.4.1 seems broken and has been fixed\nwith 3.4.2.\n\nhttps://github.com/pyca/cryptography/commit/843ada65e816a17e1b3d90b12ab6403c8ff96654\n\nChange-Id: I1dcb7d7c25ad61500de0e0c278e366976d7c4ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/239e47ff888be5d678ce3d9424b681eaf7a506a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/239e47ff888be5d678ce3d9424b681eaf7a506a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
