)]}'
{"id":"openstack%2Frequirements~774593","triplet_id":"openstack%2Frequirements~master~I53067fffc40bd7e2799b09a14b4e203607e16ab2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I53067fffc40bd7e2799b09a14b4e203607e16ab2","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-02-09 06:20:54.000000000","updated":"2021-02-09 20:25:47.000000000","submitted":"2021-02-09 20:21:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774593-openstack/requirements/constraints/noclob","meta_rev_id":"523e03f3c651f0aded43b02d6842cf815f5706d2","_number":774593,"virtual_id_number":774593,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 20:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-09 06:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-09 06:40:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 06:27:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-02-09 07:04:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-09 07:59:54.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-02-09 08:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-09 10:39:18.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"}],"messages":[{"id":"67caadbe063f897cd7c8309df5013127cc9bbf8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-09 06:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0785c865bee86259f75c51db3d59ba232d675112","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-09 06:27:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b8403742f62f7e05a765ad9d6431e0ab00cd090","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-09 06:27:29.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"99459a8e087af793d69d911e8dea8e1a7654b970","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-09 06:40:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fcbd4f18c1bf0a912b005117830025fc15c4db97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 07:04:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8c7ad5cc0bdd48378e718615fb42f91d : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4bf09e45d739312de0bd1d442893914b6fc3adf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 08:22:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f8880aac904a5993c8eab0d34cdcb6 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6894d5b8844dc7bb096d820cf9b1e4 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/420d7468d36d4abda29cb535a86c597d : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d3adf226834136bf4039fc298241dd : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32dfafd4598f4137bb28d1dd14fee05a : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e5fd35290d042c79e32dcef792016a0 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee05d3f1b85c4f7b93945322d2b5fd87 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4b1c33989164be3b2854de4c9f8d849 : SUCCESS in 9m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57b464d93bc9478f84be85390b9286b3 : SUCCESS in 7m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/272a9dbda5784a98a0a8e7a514c7de81 : SUCCESS in 8m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2352af8271414262b102a000e0a65a6a : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3743acd517f14fc58e4df7f322711b9a : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ff3d66616d94c2eac5ad34a62ac9380 : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/56b439ae278f4f7fb50bb4ed128950d7 : SUCCESS in 7m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0431c6d5e4de497cb45c54cc14ceb902 : SUCCESS in 15m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d8fd438d8ab64cb596bcfbca49dbfba0 : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5a7b34ce99b4e8f9cfa61ab1b40e330 : SUCCESS in 10m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04e59fd8615c4434b948bb225a94e183 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6224aa269fb24952a5b85247e7828414 : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4dfa1b09aca7427995a1a252eec71e3f : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b488a1071e14daf8dcc5daa00ac9389 : SUCCESS in 6m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8e220839105f4a3f9171e91f6d540d69 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45a193aba9294412b067273faa03c596 : SUCCESS in 45m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8afd54f5fce434ca893d73592eb456b : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":2},{"id":"268bd09e7e459e63257e0eb3f24e721716a76ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 08:22:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad7154fc640f94862ed663ac18ac1823f2948d97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 10:08:04.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d71cc259653421b8abc2b55f2576737 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b5dcb81a1a046ffb2ea1a5bb6983667 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b3c3357876246c7ab93b35650efb2a3 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/625643ad0ed4437abeec583cf28dc6b9 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4fc9dc49b6c4f848eeb4a1bf82934fc : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff898a1ad78c4ab78b67df5f0a5f8295 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f8a42024cfff4d0382d29b6d647d127d : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01eb6228b8b0417fa9eeb3da1aadbeff : SUCCESS in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8636ae20d89a43a88fb12bf4a760f04d : SUCCESS in 9m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/298eac59c1814fbfb20f49fadf877a39 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1a00de95e2142c0a49b5b7859905f0f : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc9c3e820e35444789ac32613349f124 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/13d392a19ef345158d7ab339407d7708 : SUCCESS in 11m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b68e5fe3a4be43f7a916d8223b00a147 : SUCCESS in 19m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/007e81bf00044ae29cfa55390dfff805 : SUCCESS in 29m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4cef78b9d8d54cc6a456adc96bf3601e : SUCCESS in 11m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77d9125a957b4351a37e3bc7ddd9dda7 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb16d7c14b8649e2a28961ca7b99a141 : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8fc11a1a217b455c938d9a26b518c5ca : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2a6ecab93bb4b50b499a5b1f02456d8 : SUCCESS in 7m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37b32599735d461aa7c24fe8bcddf6fb : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5bfbb724dbd499f9e5d8657b81f8abb : SUCCESS in 1h 02m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad963d28306a4581be83665ef800434a : FAILURE in 1h 33m 16s","accounts_in_message":[],"_revision_number":2},{"id":"63ecde88f1a43b649a17b6d000dc2b70282f3c48","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-02-09 10:39:18.000000000","message":"Patch Set 2:\n\nrecheck looks like random tempest failure.","accounts_in_message":[],"_revision_number":2},{"id":"f73f9935aa871fa38140b18e9df1bce906f71036","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-02-09 10:40:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"41e7189ad7e4c71518b8d485cf0c205f750ce933","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 12:13:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d45a5fdad6d4eb384f8785ad025eec4 : FAILURE in 1h 08m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fb3473fe918bdce1d266fe244d3d1fa18059244a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:32:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72989f6964774e1cae2656db39cf67dd : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052c4b36b3b946409033156b7388d231 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72149b40f2be4525b4c37598df7dacd5 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29de6e7751e84b5288d14617e4fe0500 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bae4936861cc488ca2652747b6292152 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d46c4765ea0740deb7cea01ee5883884 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd66f2b6128e46569f38381cadd5fded : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a69478af8cb4e51a4fdba4a41e6fc38 : SUCCESS in 11m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d483d3a174304e33a5a75fab8aa22b82 : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b31ada4e5f654e87932e939e46ca878e : SUCCESS in 10m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cae4c6b96fe1401888f46eca4891f154 : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/211fc99c7a87496e8e1b745ec155d627 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7597cb8615f647e18fe11996d050ace7 : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7ad9ecffebe440df97fda1a0e794b2b2 : SUCCESS in 9m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e2ac81bdb0240a2b3f67b17d93e2474 : SUCCESS in 29m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8b9247af82fd4ef1bb67f42025440b34 : SUCCESS in 27m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5bb52cc4624e418bb9f5abd3143e39c1 : SUCCESS in 11m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7e2c5547af749e08f6e51585839d40b : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb82f31d04cd499f97ba0b6c840ae592 : SUCCESS in 17m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ddddfdb31f29477db22eb0c8f14a9686 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d152b828c3fe4c419542d7e990d1245b : SUCCESS in 8m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd808d425ab641cdbb95ed6088fe94f5 : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9e696a85b644ddeac7aa582ec774ce6 : SUCCESS in 54m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa5535226de04b9eb7774e710a91c0c3 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"111c89d3fc28d32fbb2fdd11dda69c7339687752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:33:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ca2c38ba40ac96a0897313ff43ebdc90da45a17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:21:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecf8f5c6b0c4bbaad1252008352d05b : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b846945e6bfb4ddea9a01722992cfa34 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d395c8c05640ed8e87010fa4e982f4 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cb35ca695204ad8a0a3fefcb9a54923 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cec5151c2a63422ca9870cf9222b1938 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f211e3f8b19541ec8a8a17ffeb2131f6 : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/85658c6ac18d4a7ba5961c0241ca3b70 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2eea606d449a4f30a784ea908008e374 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3aeae80c09bb4f52ae9fefa6f4a95562 : SUCCESS in 11m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/45c8fc307f4b4919a9e6b9e5f68ecc9b : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a0ca4841c53483b80e060ba71b4f0b3 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2ecd72cad3345aaae5f96700603da7a : SUCCESS in 11m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4c0d80b7d6c0415dabd16684f90f0767 : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aba2b8ad35204840bcaf2ec3edb8b58e : SUCCESS in 21m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ff90be1cdc2448b7885d9300be5615c4 : SUCCESS in 22m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf0cb61bc9cd4dc2be1e680cbe99e4cd : SUCCESS in 13m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/62e09403df604e88a28d53abed47eac1 : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e50d5fe2f87c4e1b93f929a46c7b8dd0 : SUCCESS in 20m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ec1f89bc047c495991e6d39f10684608 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d560df750c374e07a8a2afb32f3660c1 : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4de6959a150a4d4b8b104ebbc0e8664d : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b1466c0c7a148c69faa9daa3f3a3360 : SUCCESS in 49m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1212676597994a5690cb8e568b6bfe0d : SUCCESS in 1h 39m 18s","accounts_in_message":[],"_revision_number":2},{"id":"701f1f57af2c7b5605503b3a2ee3dbe412a8d5a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:21:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"523e03f3c651f0aded43b02d6842cf815f5706d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 20:25:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6fdab5d19742e995d05a2aa3fd14c0 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4c4bd7b57937f1050be447a42db10442ce24e0d8","revisions":{"378835e99d22b93457b95cd9083dbbf713533500":{"kind":"REWORK","_number":1,"created":"2021-02-09 06:20:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/774593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/774593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/774593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/774593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/774593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/774593/1"}}},"commit":{"parents":[{"commit":"3126e526ca932a1c66e2a9db6b392c217b7153f8","subject":"Merge \"update constraint for oslo.metrics to new release 0.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3126e526ca932a1c66e2a9db6b392c217b7153f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-09 06:20:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-09 06:20:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53067fffc40bd7e2799b09a14b4e203607e16ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378835e99d22b93457b95cd9083dbbf713533500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378835e99d22b93457b95cd9083dbbf713533500"}]},"branch":"refs/heads/master"},"4c4bd7b57937f1050be447a42db10442ce24e0d8":{"kind":"REWORK","_number":2,"created":"2021-02-09 06:27:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/93/774593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/774593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/774593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/774593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/774593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/774593/2"}}},"commit":{"parents":[{"commit":"3126e526ca932a1c66e2a9db6b392c217b7153f8","subject":"Merge \"update constraint for oslo.metrics to new release 0.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3126e526ca932a1c66e2a9db6b392c217b7153f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-09 06:20:53.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-09 06:26:55.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53067fffc40bd7e2799b09a14b4e203607e16ab2\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c4bd7b57937f1050be447a42db10442ce24e0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c4bd7b57937f1050be447a42db10442ce24e0d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
