)]}'
{"id":"openstack%2Frequirements~775124","triplet_id":"openstack%2Frequirements~master~Ibe3d57e9b7712769f1374ea5ae55c08bcc7d96de","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibe3d57e9b7712769f1374ea5ae55c08bcc7d96de","subject":"update constraint for oslo.metrics to new release 0.2.1","status":"MERGED","created":"2021-02-11 11:46:19.000000000","updated":"2021-02-11 20:27:32.000000000","submitted":"2021-02-11 20:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775124-new-release","meta_rev_id":"31fa808e413e841fb72a098ef184561dc2cc278f","_number":775124,"virtual_id_number":775124,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 20:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-11 15:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-11 18:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-11 18:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 13:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-11 15:38:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-11 18:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 18:53:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f08860947e85a69327fc4d44134b78ae36adfed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-11 11:46:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c52187f487e7091fd2dfe4bbfcd59d600ca2dc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:08:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45e6c7b7ea644cf485e524f402b7ef12 : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"98451716892f83dafe67a9e04f0dfbf374712cc3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 15:38:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a71ade91ed12ccae8126b5e612bb590b29b13f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 18:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a68e69e19d4a10b6b8236865a382fd : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51401afaa730497d80097108525dc3b1 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5537e32c79704e5c85bb866b2a40b23b : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c12a9df2e2e64a8d9439e2def9724c5d : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6a85bedb0094f1885cf074cd9c8b186 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02576c5fa64c4d539779c04d6de1a630 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01aca3dc650949b6bdd010e351dd235a : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49b1d01a05ea4b04bb9d35f3c03de686 : SUCCESS in 10m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9875a441782b4248ae8cafcc5b4e7fe5 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ae790027b10442b8868294b32c2f4d2 : SUCCESS in 12m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b98d6cab8668483ab297df2cf8720149 : SUCCESS in 9m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2d33831eaca14b1abcda79ad9f9c62a1 : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1d2b1e0da67461b8b33df8a8e078cc5 : SUCCESS in 9m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0ddf7f2344644d0b84132e1786f0c07 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f6602c87b494de9bba160eba2fa587d : SUCCESS in 17m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37f451f37ea543f1a95b42126d9b7432 : SUCCESS in 26m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/21e04cec5a6f40ef8964fb0c348d398f : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9e7f6281fbe424aa739bad38316463b : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ca626da12b14a4da52c9f56912d7a15 : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fd28dd317b6a444a87f07b36235b0663 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d379b449f2b7424d846dba3f8eb9c637 : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d5d3219436874c0e837fa1a1e632c90c : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fb6bc532c4343a9a4fb0fa0b75990be : SUCCESS in 53m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd196bd0141245e49e355116c03263d5 : SUCCESS in 1h 26m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ae607b21c7e9d095d8dc6768f205fafad4743565","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-11 18:53:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d7f775e273d938035198dad2141d61bf903ac70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 18:53:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8f153927c64b9d7c155e8ccba4a0dfb03416b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 20:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67379f75e646434ea85264f37ea22aa6 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8833adc86754445c93ec5a59f6212bf3 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4349057dfde4317bbd01c8b6adf789e : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a0ff7f258844a8eb7d3a640cf98bc79 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d54db7756ca74386b077eee2b8cb9c8a : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e88aa17371f3435a85837c3400690283 : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39139438814849c396dbd82911c4a945 : SUCCESS in 14m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69c6ca7b9f06446baef3c4965319f072 : SUCCESS in 13m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4ce3588637be48b9bac425d1c395bfdf : SUCCESS in 11m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/942c8a877281404cadb79e8706518336 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/af06502a78b645cda8b585d3afff26b6 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a99e29c175d84361b3942dde731c6047 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/57e7b9b64f7c4336b77a8c5702402320 : SUCCESS in 9m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/75378af45547425585542491c6b5adbe : SUCCESS in 31m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/93bc4783a510434bb57d49f45e4f7339 : SUCCESS in 25m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cbfbb4e1f34e4c64ab387c6983e4e929 : SUCCESS in 11m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd98e9e2994d4d5084006a5fa23f43fc : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/304e7b313ad04f1295078bc98f238c46 : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/17256c5dd3e047869cc9d35563802772 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f174030e66db425aa88375c16260f718 : SUCCESS in 7m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5c37acb09f8c46b0b53ea5922e382e1d : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b61bf0523694aac87c474523ed1c80b : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86950ef821634baa913d1fe556b7c0e0 : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f78536a98361349b64e673c6dce4517632c5020a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 20:25:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"31fa808e413e841fb72a098ef184561dc2cc278f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 20:27:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8dbfd78ae64bff9d789e761d783023 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78619eb738d1b8857cda4b7a69404d8a736accda","revisions":{"78619eb738d1b8857cda4b7a69404d8a736accda":{"kind":"REWORK","_number":1,"created":"2021-02-11 11:46:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/775124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/775124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/775124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/775124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/775124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/775124/1"}}},"commit":{"parents":[{"commit":"09b8c864b4656b7a58210d75e54457452aaaecc5","subject":"Remove cap on PrettyTable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09b8c864b4656b7a58210d75e54457452aaaecc5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-11 11:46:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-11 11:46:17.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.2.1","message":"update constraint for oslo.metrics to new release 0.2.1\n\nmeta: version: 0.2.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ching Kuo \u003cching.kuo@linecorp.com\u003e\nmeta: release:Commit: Ching Kuo \u003cching.kuo@linecorp.com\u003e\nmeta: release:Change-Id: Idbc648fceb61b7a3192fbabf74f801aecdaf1afc\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ibe3d57e9b7712769f1374ea5ae55c08bcc7d96de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78619eb738d1b8857cda4b7a69404d8a736accda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78619eb738d1b8857cda4b7a69404d8a736accda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
