)]}'
{"id":"openstack%2Frequirements~775556","triplet_id":"openstack%2Frequirements~master~Ie681db93e3bec617fe5ebda586eb002eaefe6129","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ie681db93e3bec617fe5ebda586eb002eaefe6129","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-02-15 06:27:23.000000000","updated":"2021-02-15 08:04:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"837e2400f4653e48ded4bfea177cd09c289448f1","_number":775556,"virtual_id_number":775556,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-15 07:59:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 06:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-15 07:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67243f20382370c09c9fe7363bcee7399ed17e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-15 06:27:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d2a133f891b2f486268e8d554cb12e13dfb1bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 06:44:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/83b2b21c0afc45639a0eaa80cfb313d1 : FAILURE in 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4b227ffa7f36e9cefed299346d70ddf2d5c46e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 07:59:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1bd15f2cbd49188c87c413f671fd32 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f30b2fbe4d4b468d112a9707771e11 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/569fe7ceeb914d758ee202e2d34866dc : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3820de87e7e5495998973bcabc3fa75c : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17f214835555480dbd77cebee06096dc : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be28ebf6eafb4ca9a7ed5c7f45f7ed75 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/19f485fe574f417baf3c3db3a26c4f53 : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/837198bb1ed24941bc24c7d4a4d3256d : FAILURE in 6m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35e65d6832b34161afa167c2bb2ba767 : FAILURE in 6m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae08291e3b0d4a8c860cbd929898bdd4 : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/65a8369f22894cc4b51fae82e53b8789 : SUCCESS in 8m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/160ff184e1a44e79a0d506fd756a30a1 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c71b7fb4bbf24e62a185da8a6c58bfff : SUCCESS in 10m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4179bc042134452b91741189c5263d7e : SUCCESS in 9m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4eeaae1696594c5ba25f99c2513edfa2 : SUCCESS in 17m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e57b2975e3af41bcbfeabd2e221f15fe : SUCCESS in 28m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f6c2ae64f20044b4a02a67a2eb73087b : FAILURE in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8f14cde9b5b4836aaa4326547aba06b : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e6485102613401b8df0d25a2d2b1ed9 : FAILURE in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d7b333c64094214bdf5fe7904bfc985 : SUCCESS in 8m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ed492f62579412ba3e2df9d43d1a9fd : SUCCESS in 10m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96a4391f45db4ec2ae6fa6503fe2201f : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c75ad3b5f6214d5ea2c7774524e54477 : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cec424d9bb094c49960ca742c10b5660 : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"837e2400f4653e48ded4bfea177cd09c289448f1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-15 08:04:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"faa7ba977282a39dda4811d4dadc43a7d7313f20","revisions":{"faa7ba977282a39dda4811d4dadc43a7d7313f20":{"kind":"REWORK","_number":1,"created":"2021-02-15 06:27:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/775556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/775556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/775556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/775556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/775556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/775556/1"}}},"commit":{"parents":[{"commit":"fead914306d23aae6e149f3f75dbf19f563bb83d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fead914306d23aae6e149f3f75dbf19f563bb83d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-15 06:27:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-15 06:27:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie681db93e3bec617fe5ebda586eb002eaefe6129\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/faa7ba977282a39dda4811d4dadc43a7d7313f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/faa7ba977282a39dda4811d4dadc43a7d7313f20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
