)]}'
{"id":"openstack%2Frequirements~775755","triplet_id":"openstack%2Frequirements~master~Ib03b6bf3fd1ad0a8cc311023a5796760dc0063f7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib03b6bf3fd1ad0a8cc311023a5796760dc0063f7","subject":"update constraint for oslo.messaging to new release 12.7.1","status":"MERGED","created":"2021-02-16 09:08:44.000000000","updated":"2021-02-16 21:46:04.000000000","submitted":"2021-02-16 21:40:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775755-new-release","meta_rev_id":"26023f49a53c8371b3ef267eea90412fc22d57b5","_number":775755,"virtual_id_number":775755,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 21:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-16 17:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-16 10:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-16 17:09:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 09:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-16 10:22:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-16 10:28:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 17:09:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7635aec0db1c5b595da80885e8a437714cb1450f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-16 09:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ee2928eab8a04a743d4cc9ed51856951740cb6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 09:41:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/708ef64e6d4d4d21851df3ab208ca1c5 : FAILURE in 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2391a7293a5b439ae0fde8735d46221c737f7311","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 10:22:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d0eb494e128e991d8cfd0cfb400ed8c4b9fc5450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:28:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0ad50942e5485a8eedc9d13d247ce5 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb017234d8154561a3d1d16ef8fc4a72 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/513109ec3dca44fb8d48ec7931cab563 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8b7ae897bfc40faaaf8e18da231cff4 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/212c1cfaad2848f3aaacf8b88f9b09df : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f82701a45dc4c0399e3f1636149fe1a : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3defe25303c5487086869ff71f1a9478 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6b56602c8bc74298911fb51e4185b4cb : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81de76a616e84d0f807ab18d9b6b8ead : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/127d2b5e4bae4931866d73b20c240b49 : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57496b6cce424cb590ed2603837b81f0 : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9951bb6ccaa94afb8b2e5c9efbb45e9d : SUCCESS in 6m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f129f63ce41421c86b3dcefa962773f : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0e285fab33b470794b3597b85aa1414 : SUCCESS in 8m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b500cf9a8ea64a83af9871c383ffc4e1 : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/105d0c622329466d9b1380a9e5d27c2d : SUCCESS in 29m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6dd62b18162f480c861584f3800252db : SUCCESS in 10m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f39cb37c70b44923b1ccd83b30f4e7fa : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5fe517c0aad344108277131ec0fce7e8 : SUCCESS in 19m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/632b40e0e44d4ca4b667ac71c5ea2f31 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b403f30b84c44b1899c537020c345698 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/31c62cd6741a46bead3801ff5953604c : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0faad7ec48384621b18e6d101701f7f8 : SUCCESS in 49m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0874760a11047a7bd6c65313a7b31cb : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6a291d2d50f539c8a7146777743888c3b39bdae3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-16 17:09:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5df468a0f048488c52b4f61c845db8d939a8cc18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 17:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9f7ea6f3852bb9dc4c5d2fe29d87d418313903f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 18:37:10.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6abfe2c7887de101861c96756f26a4d9e77ab8ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:40:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22659f099de4c3d83413a2ac880b0c7 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f342b67e1784b6ebe7fed83619ba8e8 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b26d92157146029c89f20531238fd1 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71f7a4d513774ea6b5c81cc562c809dd : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d79754864dcf4f358acb555e9a74b264 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22b07d75e9dc4189b53861ce4c06625a : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/912f34946f9b438d9636769310f22d5c : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cde208aac2b24966ab90ec0141b18987 : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e853e99a0174803b574dfd4796b255c : SUCCESS in 11m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/78e468fa9af74110b0c9a7b2d497431e : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e04c8253b3e4b4b8cbf7c3a2b4d0fa8 : SUCCESS in 14m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83748355165142b5b4cdef2489242ba4 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/be17c44282bf4dc5b78c8e5ca70b7dd4 : SUCCESS in 9m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/38f09eabc39e487a95337d781ec9ccfe : SUCCESS in 17m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eddd147a381c4f65933a6472b7449aef : SUCCESS in 38m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/de1f572841474099b51c536a84a8c8b5 : SUCCESS in 12m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b077ed09c4148639ff5c766f38548a1 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d6f2c2ab11a41839356513d650ddc02 : SUCCESS in 22m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d482bf9d200a49a382c0d85b4097ef58 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0886ab82f6b74b589067b41b0fd97bea : SUCCESS in 6m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38c1dabaf40e4b18951b66c4208496ef : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d44db5183b444db8a3333908ead0065c : SUCCESS in 59m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7354bcf2638840dcb70ab73e2c333b48 : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":1},{"id":"53136c99b9dd856f3f439b5cb80eba6ceb562d69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:40:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26023f49a53c8371b3ef267eea90412fc22d57b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:46:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8670fa82b8f4809bfffbcd096dd0cfb : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e779cdc618eb72f970647c835df364b6ffa0942","revisions":{"3e779cdc618eb72f970647c835df364b6ffa0942":{"kind":"REWORK","_number":1,"created":"2021-02-16 09:08:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/775755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/775755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/775755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/775755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/775755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/775755/1"}}},"commit":{"parents":[{"commit":"fead914306d23aae6e149f3f75dbf19f563bb83d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fead914306d23aae6e149f3f75dbf19f563bb83d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-16 09:08:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-16 09:08:41.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.7.1","message":"update constraint for oslo.messaging to new release 12.7.1\n\nmeta: version: 12.7.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Ibd5130f49df68e87bc390c68fbdce6341c39c6e5\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ib03b6bf3fd1ad0a8cc311023a5796760dc0063f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e779cdc618eb72f970647c835df364b6ffa0942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e779cdc618eb72f970647c835df364b6ffa0942"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
