)]}'
{"id":"openstack%2Frequirements~775790","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I78844f88dca457622e8f3871737e78d4d4684ec3","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I78844f88dca457622e8f3871737e78d4d4684ec3","subject":"update constraint for oslo.cache to new release 1.37.1","status":"MERGED","created":"2021-02-16 10:09:31.000000000","updated":"2021-02-17 03:57:26.000000000","submitted":"2021-02-17 03:57:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775790-new-release","meta_rev_id":"24a4eff7466f67da7af04710b2cf1af6062f0a8a","_number":775790,"virtual_id_number":775790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 03:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-16 17:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-02-16 10:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-16 17:10:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 10:22:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-16 13:57:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 17:10:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d62d6fd20ed0eca778dcd57a62faf8f01e530b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-16 10:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"692b581e08b5f2cc8689ee0a5ae6d8c458bc3835","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 10:22:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f67d6b22ebf5947e13ea1dbf7ecf84dd2fa7114d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 13:57:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c84fcdee9f492b912d65b1a8fc863a : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdfab6363bb24220bcfffab9e8bca9e3 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90e51cd187b244e4852c4e9151bb410e : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/637fbfca67a4422eaee3c4302d0788a5 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ac976d6ad37411db5f3ffb11944ede9 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56ecd67a0f6c4cee9fbdf220e1f4e40a : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/203b3dfdc6df437ca5cd3de94d2b8bad : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be22fc0e71194e5d983601fa47eaf98c : SUCCESS in 17m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8a4adadc38a8486482a45670a43d836b : SUCCESS in 9m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef8d3fd0dc944e6e865b5396f8cd9a38 : SUCCESS in 9m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/680abdf013514552a29f8cd463201630 : SUCCESS in 5m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbf3f8f393b14100badcd1bd9572a001 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ef3442c96cfe49799a00adfb613fa408 : SUCCESS in 10m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ccb38ead6ffa4317b881585c2b310111 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e72db742272406cb3e9c3aa35aeb011 : SUCCESS in 7m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bab44b2d58c743aa869674967ad823e0 : SUCCESS in 21m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/613f48f0dda74f9a82921de9f282a7ea : SUCCESS in 14m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/312fbb162d714b2088c5c7b0838a52da : SUCCESS in 25m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/034f5859bd4d40c9b7a40d3f7463038f : SUCCESS in 22m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bd251f5d413d481ebdde481989696744 : SUCCESS in 14m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2c99ff5eb0fd48c9a0d5037b19ccb3cf : SUCCESS in 12m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9fd10f941bd94c35a43f455b85d96003 : SUCCESS in 14m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08c8a35f0754440796c0a9b1a1400421 : SUCCESS in 20m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fdf7c99c83e24ce5a514c2436fad6bc7 : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fe2ad96835c84ba49926e2825595b302 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2300f55b2f342f2a98af93509cf43cf : SUCCESS in 8m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/240767a5ede74e909f7714b13395239f : FAILURE in 1h 08m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d0fec92f43a1491ca9fdf0ebe623e71c : SUCCESS in 1h 42m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/86fd80e68c074613a5b8fea94c67f6f4 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"161a15092dfc1f85dd11b16b6afd9dc1f2c01386","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-16 17:10:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76a45f6a2e2e3afc7d597e454181c04804254970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a06dc5f58e540edbb65b9f923c761a8 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/846f5daea9e64ffd912ec422e47dc80b : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfb2b3fd446341609fff5f2fa411dbae : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d2d1c142bd24974965d990f565b74ed : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f295b9a7c6347fe8c2a388efd0c9422 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a831a78045b462eb625f49961ff4346 : SUCCESS in 3m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7bf7d8a4b72f4392b57e21fc28361fd7 : SUCCESS in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0aafecbb21814a5e95d6df0b2dcc3a1a : SUCCESS in 8m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c53148474bc0431cbfd23a2dd8274dba : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0454cd926bc94cb6848a2034d6cd9481 : SUCCESS in 8m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8674c3e4415d40579b0d0b066f2721ff : SUCCESS in 4m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e36070dbbf1b431785d0de347736272a : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c4ceec328e13465b83e1d3b3104f926b : SUCCESS in 11m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb3a40036cfb47b0a6bb72993abe3104 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e717a104c05f4bd58473f07d07b939cf : SUCCESS in 6m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/674c2fa1ebb642ab85d6d7adc597d2dd : SUCCESS in 12m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3db0a84edbbe4689abea110a881dc13b : SUCCESS in 12m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0c49d62a2a0046fda79e449bc9efbc30 : SUCCESS in 24m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c6728ef7e70242a28ff9789667ddd748 : SUCCESS in 23m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a40cc1da92e94475b1b29ac57e8dd28c : SUCCESS in 16m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8b231beac009403cb0df8bef97f869b5 : SUCCESS in 13m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/96d2e49fd9a54825a6821f5dd6579b7a : SUCCESS in 14m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c5c71b4cf6344988a65a26b6f79be3f : SUCCESS in 18m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a809bbac0b144e69aabe0fa14e832dc : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed7c5684505645ee8796babf93c7ee6e : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/99f11af24e6c48d183d81abaa7c86723 : SUCCESS in 9m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7d9d0859b084f7fa476492a60ad0fad : SUCCESS in 1h 07m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3736038509a746bdb4afcaaf07bb8d2e : SUCCESS in 1h 31m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7514b3ebd2a342858173acbd2eca0b87 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d0fd0ef65ab352a3d8204f7cd980a0dafc60ed77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 00:19:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7ab92b45731d1589acb715b4451cd7f7aba1028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 03:57:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056e00ad18a743d2a130655e98a5b2de : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3778a60bf254d2abdf09a4daa6394c4 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57ecbea529da49dca1748fb5bbcc2bb9 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/445fedb6ff5349138c8978f9149cc1c7 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cbbded7b5384550be15474b8805139a : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/903686d484794fc28a410b65178ff978 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5d21e6b8a53e4127accda8195a46bbaa : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6b7c9a277b64b208c56e1ab7880769e : SUCCESS in 8m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/18b01829844048119500c4124dd73b97 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d33dc737458466696cf2a8630e508d3 : SUCCESS in 8m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/baacf8ec3d1b4829b6dcbeaf9abc4e11 : SUCCESS in 5m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db2ac78e19a642c5a7c965b26989dace : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/104aa6e6891f4faa967e3bf851ff43f5 : SUCCESS in 11m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/da56ea86e3a441a0b2dd295410b7be7e : SUCCESS in 7m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1040682a0db2402da0c90f8daccbf3f7 : SUCCESS in 7m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c67c1235acc44cc682e9cce7dc5fc6aa : SUCCESS in 12m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ec6cd1246e104489a989dfdc7076c40b : SUCCESS in 21m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/dae0947f0a404dacb4be8d47fb7c4c3a : SUCCESS in 22m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d2442ff78f09423b8b287187527f7479 : SUCCESS in 15m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7d107a41529f468181b9efc870cdc294 : SUCCESS in 11m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c5b1b979ec4e430fbec18fe38e8e5aa8 : SUCCESS in 15m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d2cd13ae3bc440c5a0ebb3c87ce5db51 : SUCCESS in 14m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9785e9293cb4c73b7dc1c11b711ec9c : SUCCESS in 16m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9af76ae0ff2644deaeaee4dad28bf431 : SUCCESS in 5m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b321f02f612e4170baea52b7f979491f : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c00e34593a64f158d191bf6020e3422 : SUCCESS in 9m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/119895f14eda4e598f564a1953a7cb80 : SUCCESS in 1h 04m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/89ee9a6fd08143858ce4057c9b13f02b : SUCCESS in 1h 15m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/23df000a259d4c9397838d7b4a9bb12d : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"24a4eff7466f67da7af04710b2cf1af6062f0a8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 03:57:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"135f8f8256eae6170127b29c28fbd2369c6b38dd","revisions":{"135f8f8256eae6170127b29c28fbd2369c6b38dd":{"kind":"REWORK","_number":1,"created":"2021-02-16 10:09:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/775790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/775790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/775790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/775790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/775790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/775790/1"}}},"commit":{"parents":[{"commit":"72084406b4366e33cdfde3a980dc78bfbb991509","subject":"update constraint for oslo.serialization to new release 2.29.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72084406b4366e33cdfde3a980dc78bfbb991509"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-16 10:09:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-16 10:09:29.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.37.1","message":"update constraint for oslo.cache to new release 1.37.1\n\nmeta: version: 1.37.1\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\nmeta: release:Commit: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\nmeta: release:Change-Id: Ifd402f5ab863754e8a3f645fcfb4915f72cf13a6\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I78844f88dca457622e8f3871737e78d4d4684ec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/135f8f8256eae6170127b29c28fbd2369c6b38dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/135f8f8256eae6170127b29c28fbd2369c6b38dd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
