)]}'
{"id":"openstack%2Frequirements~776172","triplet_id":"openstack%2Frequirements~master~I91439abccbbaf95fb49b4508bfdc5de3b109ffee","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I91439abccbbaf95fb49b4508bfdc5de3b109ffee","subject":"update constraint for os-traits to new release 2.5.0","status":"MERGED","created":"2021-02-17 13:28:54.000000000","updated":"2021-02-18 00:34:32.000000000","submitted":"2021-02-18 00:33:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776172-new-release","meta_rev_id":"f36febe63f4681487af20be1d45bb865ab242ddf","_number":776172,"virtual_id_number":776172,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 00:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-17 14:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-17 17:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-17 17:40:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 14:10:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 14:25:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-17 16:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 17:40:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1645a7311c5601ec650447becc74ec1b6133abb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-17 13:28:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"598c09e3dc742ecbd7f9abf85eb5f0efa9b96ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 14:10:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/50884aa67a5c42b58196f89d7f626c1e : FAILURE in 35m 52s","accounts_in_message":[],"_revision_number":1},{"id":"924e24b8b852b5a169cfd280b1d21946eba42833","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-17 14:25:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ada9baa2230f9b7a22b8b3af3828b9b172305529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 16:26:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9e5e58acc047df8e4dd8f8f346976d : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246349fa32cd4f3c9389786301747531 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a852c0ca68614168af06531189aa9076 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c419b10d00564d5d91dedceca212612d : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/beafa7ff37d54819ae3996bde5a190cb : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5632ca3dad5a4ed49d1300bf4fcdd8e3 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/772cf93b708d4578a319891fe94e1b38 : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5557bc8d70954854b38d8bcb251a1b22 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cecec5a8bca74fa19f5238e75aa46f12 : SUCCESS in 10m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f9603ae30b844ae81d4781ecdbae209 : SUCCESS in 11m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/603d21f44ab74c0d85d65d92df6f1a3b : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/12b386a9804d4c558779ce6442f3c6ba : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b590ce01aa94de2ab12b833d5303abc : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7f19c1d114ae45f6b202c78eba8c4cca : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1920e31020c45ffa9aa96677e014c56 : SUCCESS in 16m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e463aed03ff34035a3ed6e9b370dfb7c : SUCCESS in 34m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d000faff50e94521a5b642ff3c4fac9f : SUCCESS in 16m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b22b1b728dd14d0c913e6a27df024ba5 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b0d906e3abf40c8874eda9e2a2abe55 : SUCCESS in 20m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7b56b0e1948c4c4aa2fc3eeef2222e04 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/930bf282feef46be9fbb727d220d629e : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0e400aef72764526a84fca1d52f80950 : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc13be21fa9c4e96934894e195ee1ab6 : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93aef19025c04125a582d94271bf42eb : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e1b5dc5e5b1fb751b2d74d19e215eb34dbcb12e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-17 17:40:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"510496748b71d85ca0c3be03aaf43d901d7010bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:43:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bed7b9740e91a2d68b9b9586064aac18464fef8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 19:26:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95189e473a4477ab986d9f3f1cff0df : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914bed3e6d204b62b5435e7c302ede8f : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41e03ea6ffff4d2e97b41fd1bd6392e7 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3784a42925a64c32aa7bb0cab345255b : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b8c988b87b347c1a96e3dfec03a7fc6 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95374121a99c4aaab1808ee49ba68fb3 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/716173de36c94f919e355ed7fea961f0 : SUCCESS in 9m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c79eefc5d9b4785991edb93792563d8 : SUCCESS in 9m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6e6b47dcf0794448abc804bee468600b : FAILURE in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6a619c821777435787b6202ce61aef81 : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de6fd7d6585a462e9bbfc2f3574b1945 : SUCCESS in 12m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb1465c88f4849a58d73d5b09d739451 : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ab2e1dbbe40444abcc23abf92c41d13 : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c7290d99b744e59867775354683e642 : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/379963ad64894292bf60d7c0ae94a141 : SUCCESS in 33m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b83b22dfb0354ddbb8909e166dec42c3 : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b827539e8a343a880bd5ac2b9dd4545 : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1ed58d64092439bb82495e065b40be6 : SUCCESS in 19m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc1d0bead59940d9a5d4ef633ba76564 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ac321ff42b34ca3ada6e71ed2847d3f : SUCCESS in 7m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87c6bd49cebf4e99b198a19bac550f8a : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa50a10f7d6345e298473979d25fb0f0 : SUCCESS in 50m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02cf12ad93074a6ebca8f138e2ebafce : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d49420d104559e53e2e51afa3f57c80b9fe62eb7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-17 19:43:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"724e5eb85ba3df76d098fea97ce16b43c945a17f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-17 19:43:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15081dd582cf24adafaec564c8e0a2abe09654a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 20:15:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f498cae74d404600a2ab9a462f01a0fc : FAILURE in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1eeaf13554c84088ac38b0eb88f2e1428865fb47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 22:00:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32299cee9cb4ac3bc4f39479da5d01e : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f073c5c9fc1415db136f090281d6e93 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c4ee84e8207471b977db928aa4f7041 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd0d9e736d704f808546a45a8c74e86a : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04ef69dac5d149c8928ff751928bbeab : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6867aa33bd8467cbfbeedd1aa565589 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59f74383afae42a0a226baabf73636a7 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/312355c3918e4828b874139c63940077 : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f5106bcf41d45c19c074121d90e31d8 : SUCCESS in 10m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/556ac04aedd84f169349e7a4e93e534c : SUCCESS in 10m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1211aed740f14187945caeecc571d21d : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/babff6fc9f534ad08164af547f9b1ac6 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c61d5a9d5754e7db9d343fe70c4e40f : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bebce53394b2491b85160cabfb5af066 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/21db5c0f627c41d6b428a9032303b03d : SUCCESS in 14m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6fbcbd5d31c34d0386aa55d8ddebaed8 : SUCCESS in 29m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c975f30c343b40989e7754f11a496232 : SUCCESS in 14m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/941e4312f6d74ad99814bd256e594b01 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c230e066d1b40f491623f42edc40880 : SUCCESS in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/689d0c3d5553461c99cf4d1541a45c0d : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f702878191e49afb5a9a91ab8a2609a : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d810a91bbad84a91ba6ca8bd38981259 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f648a3cbf02d441a8b0ea66de2f68027 : SUCCESS in 56m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d043bd3817384fa38b76759328c4cf17 : SUCCESS in 1h 40m 54s","accounts_in_message":[],"_revision_number":1},{"id":"75c567c54fca5016935bf83a6280f19f84cbd46f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 22:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f862a81080ced24abf7693dcc82b7d3b826df00e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 00:33:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a52e87eaba04bc0b98b925a4520f1de : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2effadc4790b417caae30850b9714622 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705de320975d42758816a66a35552228 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2abb6dbb9091468fade0b4a641eebdb4 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd774ec4b9fb4997bab5f5d639580ec8 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac195714ee53442c9a1eaead819096bf : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f811529895784a1b955bbed60753a7ed : SUCCESS in 9m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7cdaf3460bf6406680b9e43297e932f7 : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa63cfc17ffd42f49bfa0e1c7387e78e : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/224c9118e15e451ebd7f2e19acbca4b3 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f188d145afb411fa1b4d1573436d0a7 : SUCCESS in 7m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66a467a034ac40a6b0275649d0222933 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eca7e79a0d2148db864bcd15ff5b199a : SUCCESS in 7m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3cce8595118e42ec9d866d90ee241917 : SUCCESS in 13m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/95587adb6df74f7da19b07c7e2d33d1d : SUCCESS in 31m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a6f69dfeb14472cad98b57573b47820 : SUCCESS in 11m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ee63e43c6a74d42964f09c443ac05f0 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba9b7c0c294443d4befcd2b32f799d90 : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/10b5496e931e427794c0cf782b852b2b : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddcd2dd0bb2c4855b28bb466e31aa27a : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a3131eec1ed844dfac142ee7da13200b : SUCCESS in 25m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/000fae6cbb3d40299db4cea2b143b923 : SUCCESS in 50m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08a484d707264e7ab99624c5679df0b4 : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6bd1f4490ca272f5f419e1104847939592eb5929","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 00:33:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f36febe63f4681487af20be1d45bb865ab242ddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 00:34:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eceeb87d53a415db06307124b4af068 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e81c05f0bbf2fbe268667f4aef408b04067c9147","revisions":{"e81c05f0bbf2fbe268667f4aef408b04067c9147":{"kind":"REWORK","_number":1,"created":"2021-02-17 13:28:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/776172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/776172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/776172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/776172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/776172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/776172/1"}}},"commit":{"parents":[{"commit":"4daaba197c423946d653838dca3acb0bd800de98","subject":"Merge \"update constraint for oslo.utils to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4daaba197c423946d653838dca3acb0bd800de98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-17 13:28:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-17 13:28:51.000000000","tz":0},"subject":"update constraint for os-traits to new release 2.5.0","message":"update constraint for os-traits to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ieabbb7a9b64074102c24317eb40501fc2c0fe0ca\nmeta: release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I91439abccbbaf95fb49b4508bfdc5de3b109ffee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e81c05f0bbf2fbe268667f4aef408b04067c9147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e81c05f0bbf2fbe268667f4aef408b04067c9147"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
