)]}'
{"id":"openstack%2Frequirements~776492","triplet_id":"openstack%2Frequirements~master~I154575a82b3ce5351dba1ff0b13a60bc6b61f4b7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I154575a82b3ce5351dba1ff0b13a60bc6b61f4b7","subject":"update constraint for python-ironicclient to new release 4.6.0","status":"MERGED","created":"2021-02-18 16:31:22.000000000","updated":"2021-02-19 04:23:07.000000000","submitted":"2021-02-19 04:21:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776492-new-release","meta_rev_id":"ed9d78e4a8bc82ae1fb8e08edbdf4cb73784dfdb","_number":776492,"virtual_id_number":776492,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-19 04:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-19 03:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-19 03:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 17:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 23:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 03:00:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7815425870abe24f4051d2617fd1b5cd43cd2008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-18 16:31:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c5d20b90e8ee28e66f08cbd20a2b9c44dd88ccf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:07:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/23198f8e67af43a797e33afa55ffdd00 : FAILURE in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"534b1c9628b8c5ee174debfc89b00e2dd7655623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:42:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489ee44da6034e13b39ff03f17de8051 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87af2449d228434f9eb3e6b1594928b9 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3140222daee845ba89947a0ee098f87b : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37adde15a7ed4f24be6e1b780bf1838a : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a4aa77a79b54531b92440fbb5f7c8ba : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0db22e6f73584bd3b637c31deca22910 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9c96bf556914b8bb30809a53272bc14 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/928f2049e59a41bca518761dd6e1b890 : SUCCESS in 10m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94f73f92b6d443c1a01d106d902f0f12 : SUCCESS in 7m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c30ff78213c4d2688bccf26429565d4 : SUCCESS in 10m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d02a694fc7594b4eac84e473c34ce696 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/055ed734e60e4c0ebf96c8fcd8b53b82 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34677914d3a4427db6e43949e962c352 : SUCCESS in 7m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf0f5ca41fb74574b67c4b51c5b8814e : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/415405989ebb41229832825f05acda71 : SUCCESS in 20m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9d80e1dd1c74e5ab4e11fee88522fb9 : SUCCESS in 32m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2599a708833b4429b0d130166280dc34 : SUCCESS in 20m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1869b19fb9fe43e2839ef05031512b18 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9285a723557245469c847d6834865c48 : SUCCESS in 22m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aead4af91cb44d91874a7b51e43892f3 : SUCCESS in 7m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b51927dcf5f4c33b420403ee7b5adf7 : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71645a5886e947f09b2cefa220462d44 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5caa743d2da648dc95c9ad6ee15058bb : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09a997c296cf4d1895afa82fe51220fa : SUCCESS in 1h 30m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c0301aa6640172f163aac7af51832f3fad41ad18","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-19 03:00:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59a1187be300049d063ba51e003a6eee6c021465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 03:00:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da2ba0e7cff40e9a4882a97c1799a384a59ece73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:21:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8a5188f4a74f479704ada9d829787e : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95776b94962244d7bd125adaefe9ce31 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0161db16db844ee2a6b2af206bcab012 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a38553fa0e748d2bc70c92b0ebdf485 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea24b7fba480458f853e73cbbe6f9b09 : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abdb2925ad5244219a2a0dd69a2f8f1a : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b060f450cdb84d528e6f640dc3b36e58 : SUCCESS in 10m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/150def1839bd4a1da5bb7e857cd4821e : SUCCESS in 10m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e56f84a20a9433985f957735a5d536e : SUCCESS in 13m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a6c6204a7dd64414b89c4833aea1a521 : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f007383a75074528940b0a46a7678e05 : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/091ffdd7f3564f5083ffaa96e5d8ceee : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ad9d1d403e3460d8bbedb2f60d88d25 : SUCCESS in 8m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1017c0b45fe744649e6381fdd2078495 : SUCCESS in 13m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c5fce16c21cf423cab96e73f3e703540 : SUCCESS in 25m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46b5475d0b934bd3841aef96da62384f : SUCCESS in 11m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a3788af10ec421c88d75b4319d13ef8 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ad1fadf59f348b1bc4d779bd9862010 : SUCCESS in 19m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/966d345009ea43b78f1132bc483fde61 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6d71be49adb4213969eafc26c6b0a7c : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21ec28ce69bd42bc9364817e78dd9703 : SUCCESS in 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/098acbd34cac414391821e938b7465b1 : SUCCESS in 55m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a309295a505473eae5c0bc1d05fad08 : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d07cbba647b9356cf158b9ac26c9d2f3d72aa58c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:21:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed9d78e4a8bc82ae1fb8e08edbdf4cb73784dfdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 04:23:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e585c02c474cbe972951c8c442c4c0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b437400766ebf35b41361800c819bdb5ec2798ea","revisions":{"b437400766ebf35b41361800c819bdb5ec2798ea":{"kind":"REWORK","_number":1,"created":"2021-02-18 16:31:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/776492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/776492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/776492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/776492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/776492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/776492/1"}}},"commit":{"parents":[{"commit":"673ce445d54f8bdf792550e2d9cfdf23bffa5f83","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/673ce445d54f8bdf792550e2d9cfdf23bffa5f83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-18 16:31:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-18 16:31:21.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.6.0","message":"update constraint for python-ironicclient to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: I404dc7f7c6bff98e1de4faaa983ba0ed98de1162\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I154575a82b3ce5351dba1ff0b13a60bc6b61f4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b437400766ebf35b41361800c819bdb5ec2798ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b437400766ebf35b41361800c819bdb5ec2798ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
