)]}'
{"id":"openstack%2Frequirements~776605","triplet_id":"openstack%2Frequirements~master~I8921ee104382393a1b5a01441706ab73793d2879","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8921ee104382393a1b5a01441706ab73793d2879","subject":"update constraint for automaton to new release 2.3.0","status":"MERGED","created":"2021-02-19 11:34:37.000000000","updated":"2021-02-20 02:29:34.000000000","submitted":"2021-02-20 02:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776605-new-release","meta_rev_id":"3afcfa29720565b71ffc7ac9678148ed4eaed040","_number":776605,"virtual_id_number":776605,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-20 02:28:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-19 23:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-19 23:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 12:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-19 13:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 20:49:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-02-19 23:50:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6b5b79681fa8702de8c8373bffafc542b68635d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-19 11:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a82e19cfbca08e7695c756ba6a1769513904ccdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:10:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/21b2d1bac4dc4c02843d496b74d02044 : FAILURE in 31m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d4ac42c0fec30286e25fdc2ff7e8cb8b568022ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 13:13:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3305e0bda7b7406c9d3f83c4234e4f38 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94774c8425054debbfd0f406be8603f3 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c52425327e3f4e06b85e1d0ce4235f53 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cfbd434214b4ac292cc8bf8bfff7351 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3186ed96cbf45c196d680fd4fdd170a : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3d3e0b6b2cc4dbf98714ea6061130ac : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3908815c55154c62a3d919f9e43fdcfc : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/32ba651ea37c4a10a306a95ed3e0a379 : SUCCESS in 11m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/afcb3b58ca9441e4b45ad7e98f9f9649 : SUCCESS in 11m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64d6fc6ef10f4546a2d1d03569a8f5d0 : SUCCESS in 8m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/54893964199946ccb06d7521d18d8fb3 : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/353fd3d472ff4961b9444467568ee55b : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04b27855d580458ab937112f781d3b96 : SUCCESS in 11m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43114c0ae1c14978841b2b971dbd2f12 : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac16280a58fa4e3f9b16f10444046844 : SUCCESS in 16m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e77c013ba0cd463c99c89780ecc06d50 : SUCCESS in 29m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1cb211e76e264c669114ba75c18f4b4a : SUCCESS in 15m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b993639b25434d7296f9dfd94ebf376a : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19ee4d33dc764dfb8a65c6a4f000c822 : FAILURE in 6m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d26836dd6b774c81b652c9f0295ff547 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80cccf51fa394f3cb71930208752db57 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aabf20a43fd24e2b9cf16cccfdaff89c : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c87aa252fde408d98a00b66632e4d32 : SUCCESS in 1h 00m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61d080f4c02e4e38840f72673c8451ea : SUCCESS in 1h 31m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0f4e6d66af8e2f9a4789c0a892b3a150bff2d920","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-19 20:49:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af2fb857834092204563ffd6f24409f7973a33e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 21:25:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d0cf6dc5f3584985af6aa3a898e6890b : FAILURE in 31m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fbc5d1c060adfc97280b1e7eda998d9c825ab251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7dc3ed81a3463798fda667e341d1bb : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bf888b7c6c4f1ebad04bec0614aefe : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4cbb33593e1497d91457812568aeb3b : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bc5b5dc523a410baffa97b4dee120de : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d59f8c5f4d6b4c40b73440d35181ed79 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/beaa759a09ff469a83cb4dd6e8ee2c08 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0798c12d0c34fd898598ebe7e74ddb6 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e824a35df49643a09edfc53c6e6e7876 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/020c5f1f90ae4ba1ab51a439f8cd0316 : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c67eeadcd6ee42989e7307b94d074638 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5fbb69e087974e8e9d1ee5f11887e461 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e9ae1bfa3c44ae2ba3439340d9ae5b6 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2a3a86c235c4899b473e64c202ab4d8 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b4af72c738be40afae186a7495db740e : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7914241b0d6a4b289efa6687043a06e9 : SUCCESS in 15m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b26fc19c2b74d76a5f9b5d6baeca07e : SUCCESS in 33m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a471d61c49be4d138c53275a160272d6 : SUCCESS in 11m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c858ecb709043869d5da14efac40782 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3632f7a20d8a472dacf785a1dce0d5e5 : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fc5c98af6672447d9ec6bc44bc52151b : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87eeca7e2a0641198fafbe59c1fa5e84 : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e04dfcf5f2dd4ca1a8ed96191e2fbe93 : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed6548eda2164842a09960c0335da175 : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc3a492a56da41a795e7eaaf0ee7f767 : SUCCESS in 1h 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c04989ffbbb61d73c730dad98e7fb0055658a9fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-19 23:50:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"488d06efe4d1e2364e3ce0331a92a5ef99d452eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 23:50:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcd022ffd67cb6567f11a2cd3bc871e886cc17a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 00:03:32.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b1bb6c8d45c71e66ccda746cb432f26cd4c79d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 02:28:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2227c062f67465f9a7964455d54a569 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09986b6bd89e484bad1440971fff07df : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b709253d12b49bd8055f91ac54d82d5 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bdbc4b2da3540e69a141b02fb2200c8 : SUCCESS in 8m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e6954833ce547458a8cba5ccc4baef5 : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa2482e821e649818068ed3c4adbce03 : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9714b6114cfa4517a0d5401fa0245485 : SUCCESS in 10m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1e477bee282942bd93b7b9db27080da7 : SUCCESS in 15m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a01dfc7c0d754b5da7d06119f92f170d : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/29951cc798cc4c1f86ce26a27c651017 : SUCCESS in 8m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f094c05266e466a90b8f29671c77cb3 : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/393a5d9acdf94d7ab6a434dbeb9c15c4 : SUCCESS in 11m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0e88ac5335f4f38ab30061f0f584741 : SUCCESS in 9m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/acd4bd5fa0be4055966a03a31fe18c33 : SUCCESS in 17m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b673d7beace46d48a0750003c7b6538 : SUCCESS in 29m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b90cc71a5ef34dacbab7c173da8d7fe4 : SUCCESS in 16m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51b82986343f4f5f8fb7920c6145c6c9 : SUCCESS in 6m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf33cd0e590b4402871c51c330a5449e : SUCCESS in 23m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c7984f734d244c2684852482513277ca : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/256cd4f344c743f2b2dab3a0ba67ae51 : SUCCESS in 9m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/62837fd4afbe42308d309c54d40a60c5 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7122cd61e914de2b6b1e205e0fd5213 : SUCCESS in 59m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c75ba34ad2345a08b3f9763924c31c2 : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6830554f8fc0a2721dae98065e581cd2c8135436","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 02:28:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3afcfa29720565b71ffc7ac9678148ed4eaed040","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 02:29:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ea660d29304ca787f52962ceb80657 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7c68761a6958f31efbcef1bef0ff3fb9795248b","revisions":{"c7c68761a6958f31efbcef1bef0ff3fb9795248b":{"kind":"REWORK","_number":1,"created":"2021-02-19 11:34:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/776605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/776605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/776605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/776605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/776605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/776605/1"}}},"commit":{"parents":[{"commit":"cbd993cbff6be3730befe7dc32cf198460846195","subject":"Merge \"update constraint for python-ironicclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbd993cbff6be3730befe7dc32cf198460846195"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-19 11:34:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-19 11:34:35.000000000","tz":0},"subject":"update constraint for automaton to new release 2.3.0","message":"update constraint for automaton to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I4e42dcd53ee04c7b85517d8515781b344dee60e7\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8921ee104382393a1b5a01441706ab73793d2879\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7c68761a6958f31efbcef1bef0ff3fb9795248b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7c68761a6958f31efbcef1bef0ff3fb9795248b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
