)]}'
{"id":"openstack%2Frequirements~776823","triplet_id":"openstack%2Frequirements~master~I94b676cabf523d33c486bfe9e650b76d35ee8d00","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I94b676cabf523d33c486bfe9e650b76d35ee8d00","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-02-21 06:19:26.000000000","updated":"2021-02-21 19:00:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"50bd604f58d2ef6b630e6057b6124faf9851ca74","_number":776823,"virtual_id_number":776823,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-21 07:44:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-21 06:33:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-21 07:44:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a39dee84d0989d72e024cd616e267de11f144696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-21 06:19:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6509089364039d904e31a9ca5366b8ff15a0cae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 06:33:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2118aeb02d114b24ac80e2e153a20eed : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":1},{"id":"074613c9f3eb1fc47beb7f0d2780efd0a15f2e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-21 07:44:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d3d4769cae48ff9a22a27832bc2daf : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a172ee64a5b245abb5b59a903a9bee06 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc53963bad4841b4adee5adcabb4b219 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edd152a8c191434582ac727c791fdb90 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9eaf6dcf57134cee8e3d8fcb52c55db5 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d61480ae8c754f1591ad59fbebb284cb : SUCCESS in 7m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50819a14daa744d3add6e0cb4d67cda2 : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90f00bf844394db9af1cd312fffc0e25 : FAILURE in 6m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c98086a0683d4799b57b3f753149fa44 : FAILURE in 6m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1bd277cebf7f4de2858f1b185ef2c42e : SUCCESS in 12m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea5d5873e02841dca23638a512a6784d : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/19d9153f62e7476b9b9265c20fdb424e : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90446820909c435187f98096f0a63803 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4089aadb61374793b5249af6062a96d2 : SUCCESS in 10m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8cd79553e72846909325424515b67622 : SUCCESS in 16m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f140f762cd4c4c00a6e3bb952c16f989 : SUCCESS in 19m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ac3dc4a11424e9e982af716468e5473 : FAILURE in 14m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/864dbb621038468bacea6ef172077799 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ad3d33979964903aafd2bec229b2f04 : FAILURE in 20m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a1524d3ed7f8426faaf692519aca512c : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9c635f8933a424f98e33bb889eb5e03 : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e0d5975663a4d0baf7f4d8b580ac007 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c389b7deded24e4dbbce755b3317a01a : SUCCESS in 59m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2b2aec6271e48119e22e579b9373946 : SUCCESS in 1h 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"50bd604f58d2ef6b630e6057b6124faf9851ca74","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-21 19:00:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9c1a93213c82ad772896f73b93683cb7bb5524b","revisions":{"a9c1a93213c82ad772896f73b93683cb7bb5524b":{"kind":"REWORK","_number":1,"created":"2021-02-21 06:19:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/776823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/776823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/776823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/776823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/776823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/776823/1"}}},"commit":{"parents":[{"commit":"3d61f3f1866ecc7df693d0b2c535e30e21ee6e1f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d61f3f1866ecc7df693d0b2c535e30e21ee6e1f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-21 06:19:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-21 06:19:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I94b676cabf523d33c486bfe9e650b76d35ee8d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9c1a93213c82ad772896f73b93683cb7bb5524b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9c1a93213c82ad772896f73b93683cb7bb5524b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
