)]}'
{"id":"openstack%2Frequirements~776871","triplet_id":"openstack%2Frequirements~master~I00550c59e6820fcb15da198bd1e48358cddfde2f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I00550c59e6820fcb15da198bd1e48358cddfde2f","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-02-22 06:17:58.000000000","updated":"2021-02-22 09:48:40.000000000","submitted":"2021-02-22 09:47:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776871-openstack/requirements/constraints/noclob","meta_rev_id":"72c8fae378966783e88e50492b606bded03b8f65","_number":776871,"virtual_id_number":776871,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 09:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-22 06:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-22 06:21:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 06:21:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-02-22 06:58:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-22 08:17:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc05586c151618f682c481858398818365d1d6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-22 06:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62ef450a552e1c1b81341a17d411384f230d1426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-22 06:21:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b672f95b5806fad35dc65b4bd252f511d006cca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-22 06:21:46.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"f9e3b7f9c5b4575fced86113dda4c1ab60ef554a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-22 06:21:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c689ccb89b3ef6bfd5701f66afb3fee64fe7abdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:58:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9a138d7448624ba38acc519012ecb2b1 : FAILURE in 31m 54s","accounts_in_message":[],"_revision_number":2},{"id":"cfd7659ae6914766c887fddcea40bade2ab3b914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:17:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a94ba5af194e57b0b30aef0dac14d3 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151028690a1546199ad3cbcb8488ff3a : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa356437c0a4f3cb841c2ebfc04e3b4 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246c0a4373bd464488e7bc26ebffe426 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb5a31a1465b46e1a832f9b4288ee230 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3040155dcfdc423383662d5b6545bee0 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09ec2d3b4acc4b6ebb7beeb4efb19332 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77f98746304e4850b2115b436a4a741c : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d39dfc87def640178b06876ba3e8fa57 : SUCCESS in 9m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ce385fe27c044b68bbe1d751b5b2526f : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/390d7a1221ae4ebe9d8a519e312d53ff : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fab6893e88f446ddafdb0bd0643a5e2a : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcb763b25e5e4ff58f4fbcc1a4f723fb : SUCCESS in 11m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c97c218568af4658b9eb27711bef7bcd : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d71789722dbc4679b0417d238f401f3f : SUCCESS in 13m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c84445ad25a34070a854beff5a1d9eb1 : SUCCESS in 29m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e53f194fd41d44b3afbc8b90f5c54cbd : SUCCESS in 17m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e056abb5c504814a0570a7af0b3789e : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f646c4a1198461381cd131471cfbad4 : SUCCESS in 23m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/63da180f354e48258e2fc12a5ddeb386 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22b34837601a4d2f86b52f7f95addf15 : SUCCESS in 12m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/575e84623e794d069a48022b53fdf5e0 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e0966e8c0874f68870ef0f6b3c13d61 : SUCCESS in 48m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cefd61f3a2c341b299de8d4e681e5f05 : SUCCESS in 1h 49m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a508f7e832c30ce008c813173eb9c8d36aaa35e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:18:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf6b5a66aa81e836165bae030d7efe954530c105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:47:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd30acd79f264f14bbf5bc5631311168 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b0a89b44564586a62ae2c83b7e8942 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a640a4110bbf40f8bad281ad75f284e5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/956daf933adc4877b6131b465defa6e1 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1769d04e2c0494d984b252d425cd68f : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cada48e699504ee2bf43d3950bb36fbb : SUCCESS in 5m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ace61f47c3924df98ac8ecc18ce28989 : SUCCESS in 10m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc5d65e1a7334789a0be04ee19f91448 : SUCCESS in 16m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/161bea7b516b44ed81797090648dde0a : SUCCESS in 10m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff6df6edcfb14cbb9b5edda131498c23 : SUCCESS in 8m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab03e4d8ec73476a86510d2190bcecd1 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1c64e03d2cb44729b564ce245a765f6 : SUCCESS in 13m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c1afa466ae7451fb1beae9f313a1f2e : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cfd0fb6a4d294b33bb32c415b9b68193 : SUCCESS in 13m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8eb6e7653b1b41339fa0dae6b407d390 : SUCCESS in 32m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/16668d26869148c5bea3af9d012581f5 : SUCCESS in 12m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0dd893e081f4728bffcd17c4e95056f : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9691397a4b38497ea8c15a3d28825fcd : SUCCESS in 25m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4220ece32ebd4d8b95702abab961d2e8 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfa7c16d9cee45f49bf7b1c7bbf2d6c5 : SUCCESS in 6m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/346dcb61c0054b999629596497b0a589 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a4ab0841e944e93acf66a4dd5c40427 : SUCCESS in 53m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a24d769c860947fda7ee1bd1495fa45f : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b8263b8e791593932a740a240dd773632ade62f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:47:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"72c8fae378966783e88e50492b606bded03b8f65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:48:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a5541ada4f4aa5b0952c5c2fb58602 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7727fdc80b1a5420d144327309722bfef0b181bb","revisions":{"d050dda2d117d1330504dd68e525ab20da976a5d":{"kind":"REWORK","_number":1,"created":"2021-02-22 06:17:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/776871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/776871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/776871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/776871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/776871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/776871/1"}}},"commit":{"parents":[{"commit":"3d61f3f1866ecc7df693d0b2c535e30e21ee6e1f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d61f3f1866ecc7df693d0b2c535e30e21ee6e1f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-22 06:17:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-22 06:17:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I00550c59e6820fcb15da198bd1e48358cddfde2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d050dda2d117d1330504dd68e525ab20da976a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d050dda2d117d1330504dd68e525ab20da976a5d"}]},"branch":"refs/heads/master"},"7727fdc80b1a5420d144327309722bfef0b181bb":{"kind":"REWORK","_number":2,"created":"2021-02-22 06:21:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/71/776871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/776871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/776871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/776871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/776871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/776871/2"}}},"commit":{"parents":[{"commit":"3d61f3f1866ecc7df693d0b2c535e30e21ee6e1f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d61f3f1866ecc7df693d0b2c535e30e21ee6e1f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-22 06:17:56.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-22 06:21:25.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I00550c59e6820fcb15da198bd1e48358cddfde2f\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7727fdc80b1a5420d144327309722bfef0b181bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7727fdc80b1a5420d144327309722bfef0b181bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
