)]}'
{"id":"openstack%2Frequirements~776967","triplet_id":"openstack%2Frequirements~master~I6840ab2d2fde1d719018f8c2ac64282f17cdec40","project":"openstack/requirements","branch":"master","topic":"mock-4.0.3","hashtags":[],"change_id":"I6840ab2d2fde1d719018f8c2ac64282f17cdec40","subject":"update mock to 4.0.3","status":"MERGED","created":"2021-02-22 17:07:08.000000000","updated":"2021-02-25 11:36:49.000000000","submitted":"2021-02-25 11:35:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776967-mock-4.0.3","meta_rev_id":"e9a9be648eea0b508a61302d093b00e80ed49de3","_number":776967,"virtual_id_number":776967,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-25 11:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-22 17:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-25 06:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-25 06:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 17:20:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-22 17:39:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-22 20:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89a31d39a830ba82f5c8f7b8e6039f34b1cf8068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-22 17:07:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7d143b84e29b957a95d0a7c5c54235d1098fdd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-22 17:20:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d9c14042cae967e4de8d6a3b53a242dd0d4aa996","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 17:39:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d4af16cbb95f4df987b5cb85691360c1 : FAILURE in 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4d6cad0dc6d2834675cbad301241683ca40b2429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 20:43:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef9f5e592334bae85cb2bc4467e93a9 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5146f6a9596d4243be08113c9730689a : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79d0aeea99524a208c0b96642c50e673 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04bb124eecd447c9b4febbcbfda4eca5 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ba3a68a28f548cf866dedc19a368bf6 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/370b7327dff24387ad724030cc2f0343 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6f8b08eb9ac4dc483fcf707f7936924 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/09aab995db694a1fb092cbe013e612b3 : SUCCESS in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d03e8c50dcd4cea8d8d3c6498515798 : SUCCESS in 10m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a0153b3395542ceb95a6f2550d5da25 : SUCCESS in 10m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff5cc77ba5cc441db25c03127a93c266 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe1e9b25f1754b489636f09a56b4a629 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5819a3b81b3d4cd081cdf07bee71e9c6 : SUCCESS in 7m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c8402b84561040978e85ccb92dc27a41 : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7456053701ed4f58b8aa7ba33b3bf7ce : SUCCESS in 29m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3e98fb00d8be4dc5962bd6ab5236dd55 : SUCCESS in 33m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ad41b8307c9345b3a07fe81ef22fabb1 : FAILURE in 17m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/242977ae758849bc921908504c6b3c1d : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e5202b7036f4402b7a27e1f107e3d91 : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/559c986cb4ee4f549064c44b2b70e0f3 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b09831ff67ea4cda8523a6f59e5a1341 : SUCCESS in 7m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cda58bbc12ef42e39c8949874f8b8aa6 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed4f8b03824a447399ed6eee9df3caac : SUCCESS in 53m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/648204d5da4240698da4d0a40c5164cf : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ed0881d98ed4dde9c55cb1db0f7846f654de38f6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-22 20:55:49.000000000","message":"Patch Set 1:\n\nrecheck\n\narm should pass iirc","accounts_in_message":[],"_revision_number":1},{"id":"f967b12aba3a59800f9eb8fa6dbe0b9c86badbc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:26:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eedc4150507245b98718ef2700dc117a : FAILURE in 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5b4f31bcd5125f4db833d29ae6898fcb9d161879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 22:12:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85e0fe332b141259ffcd229af48310c : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660bbc6acf3d43bb8e6895a47d66e48e : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c93ea6b9f07c4326b06452dbbb5250d0 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e13c4fd72b64836b42131b15b09bcf6 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7effaae45bb44dcdb7e4856f5c47f724 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36a444d8d1c240d9a8bf2d620b86cb4f : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91a6f8e5aa5c4cd48dcd934ee58dc1cc : SUCCESS in 3m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a92f0d9e63a445f2b3f184694abf5f52 : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a7122dba0cd4e3b8fcd96cfa6b15071 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/500fbf713c504692ad874d8dc46877e7 : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9b87d56c86ed4e198d1bcb4e140b82df : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/75d650a0722c45be9c6cf2fcd7a258bf : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29f7a294ed074941bd85d9013ba2f5d0 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1b515de7bb0a4c6a8ebd282078a56d9f : SUCCESS in 10m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da27a92f95b643b0a198b022ef883946 : SUCCESS in 30m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5a06e04cf2ca45048f83b62d4af4035e : SUCCESS in 27m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f61acb5d8e784ad78f10118d7d58633d : FAILURE in 10m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba1e1a41af434ac4a07289194c5b3d73 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5739bb045d1345eeb91772319fdf89cb : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cee25052f8ea41acb59e250ac6a35d1f : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/620fd7a885484236b2df8cf41fa865a1 : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c00c7afa1ab54c4c83c747735befe038 : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d77535ff30a45b4a86a03997fdc0e75 : SUCCESS in 1h 02m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1de965ecbd447208eb016a863adc906 : SUCCESS in 1h 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"515a700d43d01afc2e7896b3f461a767686a6987","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-23 17:22:23.000000000","message":"Patch Set 1:\n\nneeds https://review.opendev.org/c/openstack/nova/+/777168","accounts_in_message":[],"_revision_number":1},{"id":"a27892f558230ee82033b6d736e2dae3fe81494b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-24 19:17:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9d714d0fc3ff0b1fefe8eb57a3af5c512e7bab3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 19:51:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/38dbdd7d20404b63b071cc81b41b73c6 : FAILURE in 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"67874dc29b667089b272ab86dd23710e35cf5da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 02:51:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd038d6182142f39ca45af46b10a256 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d856766b4b94e99b5ce2cdbdb30cdb0 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50652e3fd7e546fca71222d7d80d8478 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5683289558464e6da1191da3b243bc95 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65dda2e8815c46d79617d104c72fa361 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fee2369763b437896e519890b371af5 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ebe0d65f560406e96ace425dd2f58fe : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e249f3d578b42c6a9e43fd2468af639 : SUCCESS in 13m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4d3a0ba6fdc74e0d980813e67052edf5 : SUCCESS in 11m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fe7189da1430470c88db1e1acf99df19 : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aef0c0c4a74c458a9e448853cbd975ef : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/16ee926b347b4953944ffa3d6be624c5 : SUCCESS in 12m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7c62bc0b9b04c238c9a31fc705a1cd6 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e4026b9266e94a36a9fe913aa9528c9d : SUCCESS in 10m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e590d7f9fe7b48a68be4edc28faf5523 : SUCCESS in 14m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66263e6ae17d470eb7b0138c439ef296 : SUCCESS in 31m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8bb7ad5bd48441ff880c899e4a0c8201 : SUCCESS in 15m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c8f8ef70f254677a68e778b5cb8f967 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b79c51e553b64415a029989921cb9d11 : FAILURE in 21m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/45833cb8509f4878a5aa89c62d1a23b4 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc2116eba59c4ab888439979581556af : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/403a693a1a3c4b3b882dc0f7ddb33b12 : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a34c3c1c94a0437f809846f0f25d6020 : SUCCESS in 1h 05m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad998f02b0c7416a8938abb1e7e067e8 : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a9367b2398d80b5595a7a467543ca67f087b510c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-25 03:21:09.000000000","message":"Patch Set 1:\n\nhmm, cross functional this time","accounts_in_message":[],"_revision_number":1},{"id":"aee371eda392a6fbfb7451b9c2c7366c74f29e83","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-25 03:21:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"023449961e84f1e1ad8c7905bd05bf77cb3069b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 03:59:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/df0d417aa58a40828c458c66e640a511 : FAILURE in 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3a4b7ba2d90a7604aea4896324a8401db4608226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 05:23:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3feb84ee9b242c5959d9b098d287a82 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f30b2d3ee4d54330ad952f6912f232d9 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0fb7d951cbd46bda2cf06e650eea0d6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fbda39a3986467e8dff4b7a77d5290c : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/562e0ccdd1ab475cb37b0351cb457221 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae4b86f1c61a4c749b4b84a7f030bcdd : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98f5124d0c68440e88e348101b22002d : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b3d4227c928641d9904b5ba29dfd85a3 : SUCCESS in 9m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5c41200d51645bcb01c0d7c8df06dca : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ce7389ece314923bbacc966484cea29 : SUCCESS in 11m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aeec4690d6574bc4aa88f8480dc3bf10 : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d18227e413f54675a5566a58cc560818 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2887aac89ed42e38c6e968cec9e2491 : SUCCESS in 10m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/913e25be1da0488b9d250b16d67fe99e : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af60af3cde104eb5bf54e70d333b453e : SUCCESS in 13m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe4bc49c5235455d8eb436a48854a29c : SUCCESS in 34m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1bd9626d3ce44f18f6db9202d9bf08d : SUCCESS in 14m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e46350a5c314f62b12c1c5119e9631b : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6286032135243cb84b43849581cd3e8 : SUCCESS in 20m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d3d262a6ffd34f3f9a39487305819dd8 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1c07a6ac1e046f5a9800db8b7d4ace8 : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/51ca0a2a487749edbe107d12625c5256 : SUCCESS in 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbbf8bdc8e544917b7459695a488b557 : SUCCESS in 53m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf1ca439103a41c588efad2f33aca020 : SUCCESS in 1h 24m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ee97ad93c37ced49423ee4ec33d6ee5119232f8f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-25 06:20:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"686cc63c1a03733419d17bc1ebddd654cc15f573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 06:21:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f4e725e68c9035cc1ba736a1cf50da2852c68cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:35:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62c1bce7d01446c9d94da5062c5efc2 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/320593b7311d43e08e200b23f1df5d9c : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d4d463d415945e8842cbbff537efa64 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5441550c038f4790926c0427f4738608 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e4a514bedc9480f880252b7f4500153 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a814297835f74a7793fff77913ad296e : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/853a1d343c5544b5b1644d5b96871212 : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f356386d7dd448818a58e8fe65f5d57f : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/681382d8db2c4aa59161cdc5c4385f79 : SUCCESS in 10m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63d061662f8046ac81e3f78684308cfb : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6ce7e415e8f041ab8ccc83cb31dd147f : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c61ce7719b634bb289c92ab81f5bea3a : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9f10edd3b89d48bdac6096bdd7886b78 : SUCCESS in 10m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cbe5be80a39e43fea51626b7c1af72d9 : SUCCESS in 14m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/607eea8defd54fba95f0ff112d431f84 : SUCCESS in 30m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/da043670fbbb4e4aa20236576427656e : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c4eac4f103e45968efc1a7c2b36e6ba : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/060eee16844b4e0e8b46ca969f6ccaf3 : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/286413703fb64f1abeeaa699c81ade70 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/831e67cb1af543abb0477e61fa23c31b : SUCCESS in 7m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ff459bfb95404ae490189a42ed9ca3a9 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a148c154d4f4e2d8d70c3e4ba46b121 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab0a28f29a6b4e46bff2c62b3a7889f4 : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4e8f1a4e687bb981c22b83e432e82a1f81df22ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:35:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9a9be648eea0b508a61302d093b00e80ed49de3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:36:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ec32470f4d404e8e5b918c0dc77a03 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d008e5bd17580b7a803ea4908a42185565a0775","revisions":{"1d008e5bd17580b7a803ea4908a42185565a0775":{"kind":"REWORK","_number":1,"created":"2021-02-22 17:07:08.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/67/776967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/776967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/776967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/776967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/776967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/776967/1"}}},"commit":{"parents":[{"commit":"7727fdc80b1a5420d144327309722bfef0b181bb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7727fdc80b1a5420d144327309722bfef0b181bb"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-22 17:06:55.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-22 17:06:55.000000000","tz":-360},"subject":"update mock to 4.0.3","message":"update mock to 4.0.3\n\nChange-Id: I6840ab2d2fde1d719018f8c2ac64282f17cdec40\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d008e5bd17580b7a803ea4908a42185565a0775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d008e5bd17580b7a803ea4908a42185565a0775"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
