)]}'
{"id":"openstack%2Frequirements~777859","triplet_id":"openstack%2Frequirements~master~I4173b25ad03fec91b0cb76ce6b8abd5cda34da5c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I4173b25ad03fec91b0cb76ce6b8abd5cda34da5c","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-02-27 06:27:06.000000000","updated":"2021-02-27 22:30:39.000000000","submitted":"2021-02-27 22:29:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777859-openstack/requirements/constraints/noclob","meta_rev_id":"8a76de628078733f6cc99198b83b8827a83e06e0","_number":777859,"virtual_id_number":777859,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-27 22:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-27 20:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-27 20:14:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-27 06:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-27 07:22:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-02-27 08:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d5e097551f4718bcae634bf3dc5b9e84a5b0dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-02-27 06:27:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bf3d2a7c8bb3088278515a12d3b3b13a7e852fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 06:43:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cd80814429cc4416b214b3a70c804214 : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f47457bfbb08c3d9678a1953ef00cb2b0884a782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-27 07:22:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cbdbd6473eeaf746a84fb9f548ede8be2b94c67","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-27 07:23:27.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"62585df5532c760ac9510caf4dded21bf7fbff7c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-27 07:25:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ba013dc203f263c78275cf4043e6da4b4de288b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 08:01:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb146a76a48a45f9be11b911ff3b256f : FAILURE in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ebb0e3868b357cdb96e60c708fb918233944e532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 08:42:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885c59df432c416fa115e8cdfd055f19 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241c55168e62483dbf4530ef6cd94c75 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fb13e77cb20488fb3bf238793c66f7a : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b14bc47ca9b4a2a9772a91ef588d52e : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51777e9fb722424299d7d52277eabfb5 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e2c08b951584d70b4e9d4787c45be75 : FAILURE in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b717928232f84d96ae91dd177fc579f2 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fbd8b5b075d04b93b217c71fa4c0344f : SUCCESS in 9m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc8f1f4ac65249069ab89905d5335107 : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/53480bde0d8f471a815f4457aaf28136 : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8f02c53fd39d495a966efb86de0e3a5f : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3888afe221c444229c1981fbc79a76b9 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d295edec74845959f27b70033c79d44 : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/239405e13308473ea708bc04e7ea0c4b : SUCCESS in 9m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9d22837db2b74b8596fde5043e688399 : SUCCESS in 13m 35s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/21fa28ca59cc4a63afecb814df7e5bcb : SUCCESS in 4m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/df4b9a1f1db64d539c18e11fcf3fac9f : SUCCESS in 30m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1046524f91114a95a27554b4b2c9498c : SUCCESS in 12m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51d81dff05bc4d0f86d2175749eaeb61 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bdfdb516b7504b618140b3a58915749a : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d34c5f6dbc1d46c7a6151867788ea851 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13135ad807a64d369508434c8780e362 : SUCCESS in 6m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a32c5a03a9164ac484365fb0bd0bc69d : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebc182f1d7524a949215bc118b88a905 : SUCCESS in 45m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/751441252dac44fbb097e1b5bcddf11c : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b3a5996c134cfae01f5be9b25c4e8b87e7a586c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-27 20:12:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ff39cbd0ef06c32ec72b16ff283c8a9790cb05f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-27 20:13:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b51dbf7f9adb5f51a3194833da2dbad96571825","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-27 20:14:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5dc5fe571b79c2946e9fe7a551d75f1214577d3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 20:46:39.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6d61e2ad491f4e8e8715d1303a12d5c6 : FAILURE in 29m 23s","accounts_in_message":[],"_revision_number":3},{"id":"c87f226d4caf718dc0595e1f0f4e61d869628efc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 21:29:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e66793685e442eac369106c5a4814d : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c2ac2d806a426e93290a95dafff804 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f5afaf72c345ffafe045fcf76d4126 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58a78fdcea3d4fbb967a43ffc0851902 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/807c0eaee1154de98829a5f5af7c5c26 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c57935771da4521a8609b00ba353710 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4b8bf68fabb40fd9a2d3aa789069e46 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae99a094e6ad4eaaa591defd9eced16b : SUCCESS in 16m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e1aa984e37b546d8bb57c16ecf70964d : SUCCESS in 15m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/06cb0ad673af416c817ee0fa6ed834d0 : SUCCESS in 10m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/465af1832a09426b87942d956cbb0d54 : SUCCESS in 7m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/188a8fd0f4a94173879c3d5de463e489 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c88cbd0feca5412db029a521b00d6e89 : SUCCESS in 12m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2206af9c202747c79054103250597424 : SUCCESS in 9m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5b6aa06dbe5b482c8caaff04791d7107 : SUCCESS in 18m 22s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5f22f1ea61b940c3988e7e885ffb2fc8 : SUCCESS in 6m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7624fb95f2f24f95a916fc70f3f246ee : SUCCESS in 25m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0348c61dee24a9c8f1f79b3e62ba015 : SUCCESS in 16m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40fa434d24554ff0847b35e0e22c732c : SUCCESS in 8m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/745fdd6ab36b4189b84fb420c24968e6 : SUCCESS in 16m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/adfa0124a2b349f8ae8d952a2b5d0a3f : SUCCESS in 10m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66268cadde0a498c883ad34656a830a9 : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b2b73e1a4fd047a0bdcb3e0c167f6ef6 : SUCCESS in 16m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d36355dc05a1474eab6dd1c4c8a9f350 : SUCCESS in 48m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e0cc2fee81044c9bae6b8856f50b821 : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":3},{"id":"4138622f9ad350fa10c5774797d4d94feb7544e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 21:29:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"07f2dfa6037c23b3e5dc89f3098d138588d6a0ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 22:29:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de158e30d7d34cc9af65bb947dcaf9f9 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dc2c80a1e5434b82d9e945b8b0a800 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79f0e2ca2e704f1a845612437a427790 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acbb5251819443b283130f6ac347709c : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3f2895be8894b9f87f8b44b77bfe5cf : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/220dd057bc4c4d9783c79148c915dfdd : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3bf72da936b041ca914ad9ecb99cf2cb : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/210263759a5a46aeb8b290258e97683e : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e7525fa7909948aa9bf24284466dc86c : SUCCESS in 8m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d0a5fc1a001c429b8689987193c265e4 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a952dfa5187a4343aedeb9f6ecef9b07 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad49133bf59e42b7bbe65cb5a82479d6 : SUCCESS in 6m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ed72c47ae135451f96f302680eeb3804 : SUCCESS in 9m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b096ffe03d9d43bcbda9bd0325d366b2 : SUCCESS in 20m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/43ab496335404b558c94e4662a215154 : SUCCESS in 4m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/701729bb7a5a4464ad63afa223612f77 : SUCCESS in 20m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e23c49b8ca994c6ea50b4640dbcaa1aa : SUCCESS in 12m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b53b25efa3ea4a2d8bdca00a0f81dd8a : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b11d66ae4724e609fe70bb5aa8f35a4 : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a27e766daf1348cfa102e8eb0e27f8b5 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7588b147da9b45e1a2521413eceee79b : SUCCESS in 5m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a3fd7f46290c4837a1ddcf493b757e48 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ce26e66742a42c592c567a965b4c305 : SUCCESS in 48m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8a11826fd17411295d3b0b8cd600d2a : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":3},{"id":"efa187adc55b38838a7c63b8827b9f8b1d60fd36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 22:29:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8a76de628078733f6cc99198b83b8827a83e06e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 22:30:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3730a89811d6401d9591198e470265fc : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8566a9c40141afef2612496e34f13f1db052f1ba","revisions":{"e92a1711f54b2d66c061a37e8b3f3c68331c8b25":{"kind":"REWORK","_number":1,"created":"2021-02-27 06:27:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/777859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/777859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/777859/1"}}},"commit":{"parents":[{"commit":"5d3a27843a4f8d938d45bc5e376ed372b22d9397","subject":"Merge \"Cross-check with Masakari\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d3a27843a4f8d938d45bc5e376ed372b22d9397"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-27 06:27:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-27 06:27:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4173b25ad03fec91b0cb76ce6b8abd5cda34da5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e92a1711f54b2d66c061a37e8b3f3c68331c8b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e92a1711f54b2d66c061a37e8b3f3c68331c8b25"}]},"branch":"refs/heads/master"},"7d1cc955ff3c30b9a95ccabd8f22fb6c47da40e7":{"kind":"REWORK","_number":2,"created":"2021-02-27 07:22:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/59/777859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/777859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/777859/2"}}},"commit":{"parents":[{"commit":"5d3a27843a4f8d938d45bc5e376ed372b22d9397","subject":"Merge \"Cross-check with Masakari\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d3a27843a4f8d938d45bc5e376ed372b22d9397"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-27 06:27:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-27 07:22:27.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4173b25ad03fec91b0cb76ce6b8abd5cda34da5c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d1cc955ff3c30b9a95ccabd8f22fb6c47da40e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d1cc955ff3c30b9a95ccabd8f22fb6c47da40e7"}]},"branch":"refs/heads/master"},"8566a9c40141afef2612496e34f13f1db052f1ba":{"kind":"REWORK","_number":3,"created":"2021-02-27 20:13:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/59/777859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/777859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/777859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/777859/3"}}},"commit":{"parents":[{"commit":"5d3a27843a4f8d938d45bc5e376ed372b22d9397","subject":"Merge \"Cross-check with Masakari\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d3a27843a4f8d938d45bc5e376ed372b22d9397"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-02-27 06:27:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-02-27 20:13:18.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4173b25ad03fec91b0cb76ce6b8abd5cda34da5c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8566a9c40141afef2612496e34f13f1db052f1ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8566a9c40141afef2612496e34f13f1db052f1ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
