)]}'
{"id":"openstack%2Frequirements~777928","triplet_id":"openstack%2Frequirements~master~I398680e2f105d377790d9d7748d28e8e678375ad","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I398680e2f105d377790d9d7748d28e8e678375ad","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-03-01 06:34:35.000000000","updated":"2021-03-01 21:31:39.000000000","submitted":"2021-03-01 21:29:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777928-openstack/requirements/constraints/noclob","meta_rev_id":"4dafffa4c4734b8806bbf02343071c8f1d33dc91","_number":777928,"virtual_id_number":777928,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 21:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-01 07:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-01 07:30:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 06:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-01 07:30:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-01 09:00:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8a834f41035bde6a1840b702bc294ac1b6af53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-01 06:34:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84db94d216152fa63bbd25469415d3915e31252c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 06:48:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1cf4d21f3d9249ee8c149237c9f679c5 : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"df79a02d50dc7d4d0453c099b1ef08bef8901607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-01 07:30:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21cca59cc4500bc4305b2a45193347fc7cf858d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-01 07:30:31.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"bf6f8fd61c9ff4e74c883b7d099ab06d45d69541","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-01 07:30:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e179f0d44beaf177d1afb0872fc6e7628484fcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 07:43:07.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c88e8c667d364ffb89a7b9ef37ec4534 : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7cf940b18f7e6be30df325638216bccecb28c409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:00:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1bb298482da45ae81173181af1483a9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de131133045041be98b56cc82ff0f62c : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af7c35ae029f49aa9bf6bc37b31927e4 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4483f10cfd094c29b2be1c3736013594 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/870e2b676d7541698dceb6865cfcac41 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9421899310940c2a9ab94a20806d850 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b8c74dfc8e24d1682844412728c8204 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/081338d1df3744c78e8cc3f7e31dffaa : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b8be9c9793345a9adfced39e412f913 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c665ea2dfd1441ed8381a1d6206d03b9 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/226300d6768a4a1eac4879925e94a037 : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c419183b3526440396a9f3bb47b8fc19 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1267a5ab8124e25a3f735126fcf2dc0 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f374c31006a9401db67a3aad0b5d2e48 : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3433f4f4bd77440a89d9d3e25fc8f39c : SUCCESS in 13m 48s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/13643ad86dd24335bb6d4dc67c8b334d : SUCCESS in 4m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3a3c952ebd8343ffacd3460c83f010ba : SUCCESS in 28m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7baba8970e0649ada9640f028cc19b67 : SUCCESS in 12m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8fefbd196790491fb10ed9858769bdde : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79f39f8dc2a342e3a3c8a346cbbb666b : SUCCESS in 18m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80daf960bc754b8499c354765a294b2c : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64b84263abf544efa5d02dcb1be9b286 : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/864372d1652f4303b8d6d7cbbe517138 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7407a4cfded54b49b61e946a2237c51a : SUCCESS in 50m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bd411ccfb234cefbac8400b4b29e121 : SUCCESS in 1h 24m 28s","accounts_in_message":[],"_revision_number":2},{"id":"952a437d403dcfa727fdf83d07ffcf095a4395e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:01:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"54e56086e25025e6d3d3c395020137f617a387a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:33:51.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec044d77ea94da582d9910dfcacd8f1 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ba5e4120a74ff5a609b47f7a95f09a : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71a83499286841349ab98bf8284b437e : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a45d8857eff4f5c9a91c57ea3af280e : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe41c12be2dc44f299d3230ad5cdf916 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87df9f98faa84b14b246fe733d5cdd14 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7be12a79c18c4852ac5b8c7699c279ec : SUCCESS in 11m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a3f8b6d22d8b439ebe5d505a72c3fd99 : SUCCESS in 11m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6273feb4520b4524a48896107fe098bf : SUCCESS in 11m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5ec03b41c114f1093fbd3c1031acab6 : SUCCESS in 12m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fb8cc2037f874b62a919f4254826adac : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d6a7508646d4215ba6f2152cbbcc28e : SUCCESS in 17m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ee8d2a41a32e453fbce04bd01e088595 : SUCCESS in 10m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e2530a6ee1d94ecebef05d9ff75d6d02 : SUCCESS in 14m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f7676e61145a4115b8d87b7b8ca3e026 : SUCCESS in 5m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bcdb60e30f0b450db9349f771a57d661 : SUCCESS in 25m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1fef803c029f43f6abfceda93e4dfc9f : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/344710aa212c4169b4fcf74095cecba6 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39f5cd39fea64d729f3e7d24782269e7 : SUCCESS in 18m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f760e56637b841478bdf2b759164a553 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84b1cf3ead8c4ca59340e76307a4240d : SUCCESS in 7m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/84f1eeac715a4501be9bc1516af62761 : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1ea7c205d6b49dea599ddea4c44558b : SUCCESS in 52m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed8221742ec044e784f210708594a0dd : FAILURE in 1h 29m 44s","accounts_in_message":[],"_revision_number":2},{"id":"82a72285bf8f63db6d9b2c3cdebf59128c48cfd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-01 15:48:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3661d3d08e8cd5e1bb16e27c3a459b52539f10e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:24:09.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06e7793c074b4362a598fcf7d7e9a254 : FAILURE in 29m 37s","accounts_in_message":[],"_revision_number":2},{"id":"755e6270008a2265e3baf42e7104c6aedb68d846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:10:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1757b6d5fde422894cafdb0bbd18c99 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b6719ee0c2409ca0a536d4c1f91160 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50e40b172b8249aa96f83e165132fda6 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8c55ce3b77433a9dd2d47fbbdf995f : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/385d861a910d45729c20e3324cbd8777 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dc1f3fe1f5844488de53df288221244 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1bd2389e30a4643a28a19c4f76a48ff : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f72bfd0735043f8ab6b3a73e7111cf7 : SUCCESS in 11m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a7f0200e9b20444da83a5b74950103b3 : SUCCESS in 11m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f0ea77b25aa4263950e2c4b9945c66b : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bbc6986c6d504235b888ed34528eb45e : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/607f68f9eb924bfb84e0ada5fa5a38aa : SUCCESS in 13m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bc8e705160343c4aeac0b1506c487a3 : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2dbbcc5296674126b6c878994522ed9f : SUCCESS in 12m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5586478ba21e4187bf0445af162cb598 : SUCCESS in 15m 18s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e9943ef6323447f0a95976bb9d17d039 : SUCCESS in 4m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a59b3b759ed4490ca9794b836fe9f69e : SUCCESS in 25m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/341c35f084a14c3daa0eb349b3cac10f : SUCCESS in 16m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f5f58a1a00e4cb6a048b4822c5c2caf : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dd4cb7561ca428eb2f7906007a98285 : SUCCESS in 22m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/29a367b48c3d40569d8d20bc5d8246a0 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df01ea7cfa2a43b4815fa5a0fb5384c7 : SUCCESS in 10m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1069286619cd48d3b297daa192ca9fa4 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aeae71980374659bd54a3e61e899090 : SUCCESS in 52m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c7fec9ea3224db3a6347e5dfaec1ec2 : SUCCESS in 1h 44m 37s","accounts_in_message":[],"_revision_number":2},{"id":"9ad7e29f5dfd24e7788343d557dd8e27ba269845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:10:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc0fdd10ec7c5c9673e7efa5061aad8a55b24e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:29:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca050ae6aff47c288a16102d06e5520 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ced8c1e5004c8f916fa980fbf89575 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/403abc1bc7ff49ae95b796e1f19d8203 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/912ff12783d248bdb35f9feb3c6a7906 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/591c0e4ad0ff4043a17c79fa3192aec7 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0b1405736ca416792559fd17a9e6448 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a6793ce41974845a1f07c4f7c17d48b : SUCCESS in 10m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9774a44e174849a181dc81c73d7d6b83 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f24d6157c9f34da8aa7500356989d27b : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a916275948404343b67eb076b4270fd6 : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/af2a383e23134f52bd9d04d2bc873140 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/291344e091a6496cae89b781fb672f23 : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/381488d4325349659699eb0bd96eb58c : SUCCESS in 12m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b98d9da0a2c14742866d2c848704cbbb : SUCCESS in 19m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/97cbf897488f45bc856b8ce4e4a960d4 : SUCCESS in 4m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1dea063df0ff407f868052b9b0208ff1 : SUCCESS in 20m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0746b96e01c4a6b89d4fc37f0b8f728 : SUCCESS in 11m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/028608a4f4794d4b88fecc0d6b1aab78 : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbaa57be66bf4e81953c069c2a79d43c : SUCCESS in 22m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e187202ddbbf4422a837119cb8242da8 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa68141632ed47bfadc82f24ab361da2 : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d7ec605c7594fc2b5c57fe59d047938 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5de0c84c9794ecabb28475e5af8874b : SUCCESS in 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76eeb0e2f1ee487f9284f96b12421cf5 : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4671cde233504c3346bfb5e14f03671486688b45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:29:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4dafffa4c4734b8806bbf02343071c8f1d33dc91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:31:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cb9af7233d48ccb781feb330fa92eb : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fb0cabbea075eb9a61087c2a9ff11dd5475f641","revisions":{"e9508b9aa788e7b22ba6ff6da51812f456aa8a09":{"kind":"REWORK","_number":1,"created":"2021-03-01 06:34:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/777928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/777928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/777928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/777928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/777928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/777928/1"}}},"commit":{"parents":[{"commit":"8566a9c40141afef2612496e34f13f1db052f1ba","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8566a9c40141afef2612496e34f13f1db052f1ba"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-01 06:34:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-01 06:34:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I398680e2f105d377790d9d7748d28e8e678375ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9508b9aa788e7b22ba6ff6da51812f456aa8a09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9508b9aa788e7b22ba6ff6da51812f456aa8a09"}]},"branch":"refs/heads/master"},"8fb0cabbea075eb9a61087c2a9ff11dd5475f641":{"kind":"REWORK","_number":2,"created":"2021-03-01 07:30:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/28/777928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/777928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/777928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/777928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/777928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/777928/2"}}},"commit":{"parents":[{"commit":"8566a9c40141afef2612496e34f13f1db052f1ba","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8566a9c40141afef2612496e34f13f1db052f1ba"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-01 06:34:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-01 07:30:05.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I398680e2f105d377790d9d7748d28e8e678375ad\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fb0cabbea075eb9a61087c2a9ff11dd5475f641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fb0cabbea075eb9a61087c2a9ff11dd5475f641"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
