)]}'
{"id":"openstack%2Frequirements~778461","triplet_id":"openstack%2Frequirements~master~I023c94f95f6b379286fea8421a43e9936f3de779","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I023c94f95f6b379286fea8421a43e9936f3de779","subject":"update constraint for castellan to new release 3.7.1","status":"MERGED","created":"2021-03-03 13:46:53.000000000","updated":"2021-03-04 08:25:19.000000000","submitted":"2021-03-04 08:22:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778461-new-release","meta_rev_id":"90bb23f08703321d877218f496b4edf300a1f927","_number":778461,"virtual_id_number":778461,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 08:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-03-03 13:49:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-03-03 20:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-03 16:16:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-03 20:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 13:49:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-03-03 14:34:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 15:04:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 15:27:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-03-03 16:16:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-03 20:46:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"19c782f36d90431c7aae4e40d1c8d34c53dbce61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 13:46:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fe50f81d8166a53d59a54d785b2a5626a9debc3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-03 13:49:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"970007a9d80de92cb7c35efccb753c32f6fa966e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:34:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1cc1ab750564475eab0bbba077369e74 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"04e13ed9e782c2ee431eb917f5f459882e4fa60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:04:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead1a8ed9b704a388d6fd91b834f5ac4 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11014c284120407a9967414418d0fc8f : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bdb118616a04a7c9918c2dc728009d9 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/954ddb6103f34a58b9113ea91e37a27b : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a932f31f8ff546b3a2c90df026528437 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1faca36b511640cba1b9ce657791cf88 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6e3547e611a4b94893a77e6c9d15da2 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5eba5743e35c43fcb6760cb08c6363d3 : SUCCESS in 10m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7014e32589ae4d3c86e397ed32a1e0a6 : SUCCESS in 10m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e07688f2b6004c3bba8957ca72575fd4 : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/feffa0e4b08248948a1a2606b30f95e1 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3283e5394ea84866b0fd6dc2497c3c57 : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c854f9c9a8d0464883e05e6bdd5ceb07 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0d6a7577ea1f41f481c18d323607eac4 : SUCCESS in 13m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/56f8c85220744529a38a525255eb01ce : SUCCESS in 16m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fcdba06f52074557bc7c875c8db6a943 : SUCCESS in 5m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/af79ea05cc25434db0aaf04880ceb751 : SUCCESS in 24m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e7b18aaefead4544bd2264eb0745b181 : SUCCESS in 12m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01f9f12092a042cbbf68924fd67409ac : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f697e36d2b1409bb4098911a538a402 : SUCCESS in 19m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4c7abea1f4f5473da59f95cf025a67c5 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b980ae0c8c84f93b4b8276e9eb7baf9 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b5f48616d3534559abb0df7ca433c4cc : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b04e9bd780b4d519eb7ec852b055d2d : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd5395152f374f9e957ce1625ad5fab2 : FAILURE in 1h 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"65eb6d2c8b0de4ed138a0600397ed1a0d4831a36","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 15:27:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5990d79ce55db207cffb55341a8ed4941b7d5a17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:14:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/16d1798a9d0f427b8438f74e08b6882a : FAILURE in 33m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6efedca44aff8fe8893e0050124373b283b34698","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 16:16:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8706a62429dd98c27414810ec2afe4417c004def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310ce781bf8f43e386782825c06d06c7 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c8e3e0d06b4b6fbb77218ef7a55fd1 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17492483b2c042f7a03c683dfb894d9f : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b3b743a6db42dc9d13c786d697e32c : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84dc2f33af7c47d280ae52f81d297770 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28ee94b1c37b4072943bef36c08ec923 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/425ab48d6f8e45a29b603f0882b38a1f : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/879d0d322bf64c0b8b249eb00678798c : SUCCESS in 10m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c4f1a6f6f2114e79834678566b7f7c1a : SUCCESS in 11m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ba6ae44014b415f8181b3f1e188dc44 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4c93fae94a04aa195b6cf84bc5c926e : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/62ac7d26555745a592a17c38eb33fa1a : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61ef7a4c2fb9492a836b312fc4b4b74e : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1718d68a4075418889920a4a9e5bc0d9 : SUCCESS in 10m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ab0d051cd864432a19b6ce7279b9477 : SUCCESS in 28m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c364b9678e5f460181772cface82ff97 : SUCCESS in 6m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a87e4869c95a4c01a94e14021ca8169b : SUCCESS in 32m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0c7cd3a034b416cb201e39efc329dbc : SUCCESS in 17m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3220b47cf5dd4d83a0b092c25bbe137b : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b55a52df21aa4853810307474d499de4 : SUCCESS in 23m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c00f8e5d0ae945a1a98b2f919b2dfa6a : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c73e3dbc33d84a64a56afab286ec273c : SUCCESS in 11m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09fe662f866f48808b7c1441ec51ec76 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef0c66eb3dd345f49da4113f5da9cd4e : SUCCESS in 58m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fc2e51c65124b16919af28261fcad2a : SUCCESS in 1h 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1afcdd913e3ee5df4ab1faa07c73227b5ab26f48","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 20:46:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c6d0dad49be97a6a602e2e55d23e428fecc7cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:57:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22821e07ca4cbb7fb7e50cf7c34f53532482bc9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:17:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a2fb65cc774b658aa4c8300d5d3937 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d218eea0e0746ab8c67a2ce393a79b5 : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b22f71fc20ed4d84b6dcd5ff2740f937 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc4aa46d7ac243e6a858e17cad12e507 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85242c5142f04f6a8748ea16354c7a9b : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4662c94a4e3b4b2c8383b63ccf0e4425 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3a96dca2ad74dbf90d4a41592c3643e : SUCCESS in 12m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94bf6f3238e24006a75e0dd967762512 : SUCCESS in 21m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1e53c56ba5a24111b522fc8c0131684c : SUCCESS in 21m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/654ee27fcc014cd6b3ed4edf4bdcaf18 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1c1a5f9b4004a2687da32e1dbf50c49 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f13a79b04e9449e0bb5a6f6182109607 : POST_FAILURE in 13m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/510c058b50854789946cc77501946159 : SUCCESS in 12m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/07762c2a791241dfa5bdc419c023f0a9 : SUCCESS in 14m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/112c16d7c9c44f278461ffda78f3b936 : SUCCESS in 5m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a6887871aa874b12b70ba90531d3fdca : SUCCESS in 32m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2bcd954a4e29440b8669858eec2a6ba6 : SUCCESS in 13m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1fecb07240c4806958c6056f7faef7d : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bee1faeda95423f9921c1ac7ebf9469 : SUCCESS in 17m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4ed2d1eb6ed5416d9f092b0bec9ec47f : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/464334081ad4466aa4a9e14f97acba0e : SUCCESS in 14m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c9d6aedf7884e0682f383a36c9c5d88 : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59ba488e83d24b6f87f3e5ebc648f50a : SUCCESS in 46m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a3b38e667f04f999732b13503201ae2 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":1},{"id":"955125ba707da057e79650f35017a922ef8b55f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 01:47:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b08a22a88c50c086296b1db62a53959df1a06135","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:24:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9578416f9b1747b39a6d287ec7ec53c0 : FAILURE in 31m 25s","accounts_in_message":[],"_revision_number":1},{"id":"03c445cac2e6a18bd05d4b9a1846924d20d7716a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 03:15:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3450178945b49b09492035e64d4c6ce : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7816e81ddcde4983b955d7a126ad06d9 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ca24f2562ce4f399915fa6195d379de : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b619687c1834b2b9d3343b8a4436f62 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/787285d0bfec40368f2834027d7056ef : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/324ecc1e774a4b32ad502cca1cebf281 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47d9a82afd2446b186dd180130bc99ac : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b979a3120a2743b68ac4254bcfdc2809 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/45d912782ec74fbfbc7b8df3f366ccef : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ad35c8be292497893594a5cf10b7443 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aef130d5ad664469a4a42106fe925d9c : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7222dc80b74844e4b7a7da8cc60c6b17 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f68cc3be5bc4593a4a0a9a87c57b099 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7a87ba45786a48c68c60dde5d2009485 : SUCCESS in 10m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/708ff650770f4fb19e1e8969c5d6e7e6 : SUCCESS in 15m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/01bc31e520e64f1e87543ce1d68d888f : SUCCESS in 5m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c92081d8351c4a0db226c9bdaa23c552 : SUCCESS in 30m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f4e4f6181d25466c9ae61aaba0802c37 : SUCCESS in 13m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6529158a2d5f4266ab2454b4a1116e3b : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24a5196cfbf74b84be6f71405a421c43 : SUCCESS in 19m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ebeb08a873624a759dd764d28b655316 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1285b904351428d970795a1d025d63f : SUCCESS in 7m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9605f55b33b24befbe4d6ec3a234227d : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/215b18056e2341dea4b9afc43906443c : SUCCESS in 57m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa853d556d0e4bc3931e2cdfb30e978b : SUCCESS in 1h 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3b359cbee7bdfc71c7659885f4d6bdc074f6fc1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 03:15:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcd1986d168312c2ca22726ba8f2c9a9e3e88269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 05:10:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5bc340a504c44e5b4150c40b9ae9883 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aed4fd664c64fd58449bc4aade39c84 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58b8f2d543cc44ae816ccd4898b1979e : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75c8a0a9cc1049168299114cad9bee07 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aae15be490784bcd9ec7a213390c5c59 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad94455285a24da7a25b7c555d4bd117 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a9829b7d1094210a1af191a3ae638bc : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d3a15557a984e76b905ef449619e8a4 : SUCCESS in 12m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/29bc80ac7424414ca14b8a4ec34c6a5b : SUCCESS in 12m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a6f5f0a3fc594dfd8ec4d5178b21edd9 : SUCCESS in 6m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bf8feb0a541d489c8c9a3946fe1190a9 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5843cd1e36845a89a7acaaba2424c1c : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9918a5d49e27476f9bd5a540b809d6a4 : SUCCESS in 12m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/179e74645bc74205ac518ff9af52a2dc : SUCCESS in 17m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c33f9415fd6d4a478690684d8fc1e002 : SUCCESS in 4m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dab1a00949dc4ba78945462c08b3abdd : SUCCESS in 21m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8287ed8467494f22be93d4c9ad31a30a : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c54c8dd503644fc8dcc2b98026f23d9 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1af6f100598415c945f4bb10ce73996 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df8859fef4354154be0cccffb80733c4 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/670db5e7f14e40ec80948ec9b7f38991 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f1e19d041ac4d2e9f51187e48b9512c : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b146eab3f124721b52dca3206fd1049 : SUCCESS in 50m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d962349ec04641399d17d8559de6ce0d : FAILURE in 1h 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a033dc53d3ab71502917fc513890b299e9c11b69","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 05:11:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22b5df6ca446f074ebe296c56233a3afa32ef0bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 05:49:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/40798122058a4286b3ca9e94a77245da : FAILURE in 33m 42s","accounts_in_message":[],"_revision_number":1},{"id":"39754605e408db863500ef0b1cafad33ab407e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:48:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a729571dd2254cf4b2a28c7a11ca0cab : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8d8f634b3dc4cffa695edf8bb2866f4 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9499895cc8b4ba9b8cc810b680b1db5 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c035060ce4740b284e905ca78374f2d : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ac8e9f332ac4d8d9711b6f0ebbf3330 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/920d3dfa1406421bb1b3b0c54e914339 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/715112df618745e9a1b3ed7fa1c19b41 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/78c11ae67d0f433ab08dfd9a64196414 : SUCCESS in 7m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c6f5155f42a468a87396c8d1aef5b31 : SUCCESS in 10m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd5cf2a902f44bd3875d151ac2667050 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3b657fa771184a638d32501285e788ca : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f9a389f503244c5284d6c40cf34d865d : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/325000e2ccd242a3865a773f37428c26 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3decb16edf94c2ab543b1ccff77da57 : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9612f825b5b64c20a3a4cf1e7eadc3b3 : SUCCESS in 15m 27s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/bd854f43b0954658b466e5e24148fa03 : SUCCESS in 4m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/535c9aebb5434f0ab578762388902a67 : SUCCESS in 24m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3a2eef4eb33840c2ad601d617e1a8b29 : SUCCESS in 18m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b87e3a7a5b647499a2706fe91e53089 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97b96320a0bb4af99ebca7fc1d6fac13 : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b424f0e87bfa43c49e89db23c052a7f7 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb1df5e169ee4ba9bad31efa5211a801 : SUCCESS in 5m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c6f589432464e03973de1fe2588b396 : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc30e80a1d7844ddbbbf031bb66c401e : SUCCESS in 1h 01m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f737f3acec444cd48ff5f78212d17cca : SUCCESS in 1h 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"903aaa6e3988d1185bf2638daa39d4fdd115995a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:48:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0702a2256bd9bf7302f4d260a7281fb637930760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 08:21:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5175b5c2a442a38968ff6525ad84ef : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa40c7704d0f4319bc945ef58294f9f0 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0e9c6a417fa449dab70b82644c1a5b3 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c700bcd0707426b904d14c7bf5f04ff : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/460be5221e734ec3a63beef93e4f2451 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bee689e07fb740508345bdc8295493b4 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/89cad71f93354e94be6e62955919e2fb : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8110537f5e544782b5d33d70fd3e96b1 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fb97715b4a35477197cfb91ecf2fc619 : SUCCESS in 10m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f010c5d92afd4a52b5cff3f3874858a2 : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/064dc60c351c4621b69e666ac998ca31 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de7803cfac264c2c97ccdadde2a5ea8c : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/21c39774d17b492d99d76601f6a21a94 : SUCCESS in 8m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2bcb03b1ff904fb7be19cb323d4f27d9 : SUCCESS in 13m 41s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e5db37c699704cfbacb28a3d194f3c3f : SUCCESS in 4m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/50880290151447cb9b0f6c8bf477964d : SUCCESS in 32m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c8f90230a0a149b49d60f34024a37c93 : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3b1b19369be4014985e1adb04e71e47 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6de096823df40d58c6a7032111c3841 : SUCCESS in 16m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/596d56cc97ed4f0caf99ee20aa1fc5e9 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9097f366684435c97b99dd4fb907fe0 : SUCCESS in 7m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aae3f1f379fb49f98fc6ea8df7a5967f : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f4912ccf3524746b773bdb4c15732c8 : SUCCESS in 59m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/982283a4d0a845ceaeab5c4bd89118e8 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2dd49202c810737bc32160f133be4365a6165bfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 08:22:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"90bb23f08703321d877218f496b4edf300a1f927","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 08:25:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ca90a074234b2aaff253b574dfcdb3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55b1f849fcfd6ee88c91f2ce2a89cbe3f007e7da","revisions":{"55b1f849fcfd6ee88c91f2ce2a89cbe3f007e7da":{"kind":"REWORK","_number":1,"created":"2021-03-03 13:46:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/778461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/778461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/778461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/778461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/778461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/778461/1"}}},"commit":{"parents":[{"commit":"32ced57cbac8c2713f982bb094b9b2ab997ddea3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ced57cbac8c2713f982bb094b9b2ab997ddea3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 13:46:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 13:46:50.000000000","tz":0},"subject":"update constraint for castellan to new release 3.7.1","message":"update constraint for castellan to new release 3.7.1\n\nmeta: version: 3.7.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Id73da4b9db2ed57c06ebf9d541f8baee5b5003d4\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I023c94f95f6b379286fea8421a43e9936f3de779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55b1f849fcfd6ee88c91f2ce2a89cbe3f007e7da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55b1f849fcfd6ee88c91f2ce2a89cbe3f007e7da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
