)]}'
{"id":"openstack%2Frequirements~778465","triplet_id":"openstack%2Frequirements~master~Ic47b5026ff024bfe0214db0c5c57c254878c7f7d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic47b5026ff024bfe0214db0c5c57c254878c7f7d","subject":"update constraint for oslo.config to new release 8.5.0","status":"MERGED","created":"2021-03-03 13:50:23.000000000","updated":"2021-03-03 21:41:59.000000000","submitted":"2021-03-03 21:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778465-new-release","meta_rev_id":"f85b3b604f7a39a956c3032dab7a26dc4d7647a9","_number":778465,"virtual_id_number":778465,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 21:38:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-03-03 13:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-03-03 15:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-03 16:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-03 15:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 13:58:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-03-03 14:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 15:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 15:55:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-03 16:17:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"007a39e9efc0c27ce6f0b2bf7f602235652429e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 13:50:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"864a6a08bd385ae58ae46c4015cee5d91bbc4341","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-03 13:58:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"69c08fb89c4aa26003966f4a139cd5f16ead4af6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:38:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f4a82ee571b64232867cdddddc9413f4 : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"371a3342f6f23213a8e27cedee77379f9390d723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:52:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4867556ed1472fa3e069ff165b28cd : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39eddd233354320a65f6082e4db8028 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6515e92780ab4bc5bb733c3ecf267619 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/419d38e382ba48299b5637585b33f1d7 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/074592cbd4514c5888b12bfcfd595902 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96356bb63409469a945916759c60e4de : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecfd8e136048425fb55ad9dcddb317c1 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f1cbe1e409f64cfbaa5e609a9753c246 : SUCCESS in 10m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4cb103e580eb4452a9fc6b58c3299457 : SUCCESS in 12m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d349433d5a348ef8573211e8a86b9ab : SUCCESS in 10m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1e81a8d81be4cebadf5214f219f4c04 : SUCCESS in 7m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/00d653d9e1424e09a57f704f456a3784 : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e76aebd443734856b7003fcfcb346fb0 : SUCCESS in 12m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f1137e790bd649cb85791f95e806fbc6 : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5f6e3e82b1194025bec64e3aafb90d0a : SUCCESS in 18m 02s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3883fe0b291c49bcba8a4d0aff7f7ece : SUCCESS in 5m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1dc444a031774f8e88e403582c98801f : SUCCESS in 32m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/86053a892f404317a898de3f92dd8f43 : SUCCESS in 14m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b56e367c29934104a4bc05db3b877dbd : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff31acfe58dc4e66adf3b9cbf365b0cc : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/602f1968157d4329a714cf0a972c25e8 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26829113434a44e69a4af00097087750 : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/98e85eeab42c4871bd8b0085eb145097 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a43368aac5b04c6a9dd3f0a4492a3c41 : SUCCESS in 1h 01m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02187053824746c096a84f26f3fb396e : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6aa26af430250d436eeeb91bb6f8bb16dd630218","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 15:55:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7202e01d1fd304a2830f2faa7bfce94cd94969d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 16:17:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48e6846424108664330513b2a5d6bf2cd2069407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 17:06:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8aae6220737feeeadd11ae996b576719810ff90a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:38:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa70d3593b94ca0bd15d896a25b7760 : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264fdb9687a644baa2f5ffb4a04a300a : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db8dbf96d22b41c0b921c18b84215097 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54ef6e99158f409981855267762e193d : SUCCESS in 10m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/580fedbece4241799c7ea1a92a992e91 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a3d89373ba340c1a1992e8bc72d50b7 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/673cef2f374b431597e9af1e3bd26ce9 : SUCCESS in 11m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee9f72f674cc4117b8b0aca5ffcfdd29 : SUCCESS in 11m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b474232409dc4de7862aa73ef7d42608 : SUCCESS in 13m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6eb030444d72447580c2e49a9f6ffb4e : SUCCESS in 8m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d9528ea451843fea07b136360a9ac25 : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7aeb9cdb33aa4fd18df80fc8fc1828b9 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5a62e6265aaf4988bd012970ff0e9343 : SUCCESS in 13m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c785ec7ff76c468eb264e8c0e851611b : SUCCESS in 14m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/aea60f10920e4f05a34c0c586c39fda6 : SUCCESS in 7m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ddd9cb5a57d449bf86f863b6f91f1e06 : SUCCESS in 33m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62f0c6f4d31746cbba9d4afece4d857c : SUCCESS in 15m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/226c574da8604ca2ae897e99f0517ee0 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3bf1649b8af4c4cb85469d6e574891c : SUCCESS in 19m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b5330eb546014ba48eaf4034920b7d81 : SUCCESS in 8m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00af0da268304a82a60b2e3767da8c03 : SUCCESS in 13m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6dcfb40a38004db0b72d30176d29873a : SUCCESS in 15m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bf7b2004a7e487499c49d4f2885bda1 : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4415d2a714404c2b8ed8c33df4b296d8 : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"07faca25d44256ba5990c404bdc77f35accb8582","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:38:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f85b3b604f7a39a956c3032dab7a26dc4d7647a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:41:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8bd19e437cf44a28adf0e08f0e58de9 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae83dc0872541ed23f24035b553e87b9ff170ec5","revisions":{"ae83dc0872541ed23f24035b553e87b9ff170ec5":{"kind":"REWORK","_number":1,"created":"2021-03-03 13:50:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/778465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/778465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/778465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/778465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/778465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/778465/1"}}},"commit":{"parents":[{"commit":"32ced57cbac8c2713f982bb094b9b2ab997ddea3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ced57cbac8c2713f982bb094b9b2ab997ddea3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 13:50:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 13:50:16.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.5.0","message":"update constraint for oslo.config to new release 8.5.0\n\nmeta: version: 8.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Idbec8dc7663db40f3cda029e5b588071edc83f3b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Ic47b5026ff024bfe0214db0c5c57c254878c7f7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae83dc0872541ed23f24035b553e87b9ff170ec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae83dc0872541ed23f24035b553e87b9ff170ec5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
