)]}'
{"id":"openstack%2Frequirements~778466","triplet_id":"openstack%2Frequirements~master~Ib790097b6ce42611258c1ced57fcdf51183cd70c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib790097b6ce42611258c1ced57fcdf51183cd70c","subject":"update constraint for oslo.context to new release 3.2.0","status":"MERGED","created":"2021-03-03 13:50:24.000000000","updated":"2021-03-03 21:36:17.000000000","submitted":"2021-03-03 21:19:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778466-new-release","meta_rev_id":"e76ee52a6b97ec74294b58f39d24a15040276f3f","_number":778466,"virtual_id_number":778466,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 21:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-03-03 13:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-03-03 15:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-03 16:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-03 15:40:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 13:57:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-03-03 14:39:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 15:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 15:40:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-03 16:17:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"fbd25dfca406f6d450da13c6ecc4db636edb8142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 13:50:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15c47c4a83637ec9c89ee7441ab1a0ada47d027d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-03 13:57:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c13a7d8bec0270f405f3a1d4a7e421332f11315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:39:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2cb6c0a2833346d3944f180baeb2fac8 : FAILURE in 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7602c0a7154fb7293e75884fdd6ce95f1e14ec93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8781c80fe28f44dfa1e7e51ce9d67a15 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5bfb71d1f44c3ea3224d20c0d35a98 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e97d1b9f62404a900f4c9156115a0a : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db5067ab8601493d9bdc2e0ac3d0265a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3685603837bf4cac815a4d758d1eed6c : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45676fdd9abf433aaf616a9df82ddbfc : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/078696d8738549c6b81ab545da959ecc : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bae661d766dd49a69254763027834605 : SUCCESS in 13m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4e339e067e04888b35b926cb1de3a6c : SUCCESS in 14m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cd9db23d777d438498d864b8ccd769ec : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/91683e183b954e8fa46c878a2697b73b : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9aee5e9e4e734183858b5aa780695669 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a84bc718ab242d89f2f6eee5c5db8fc : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/df6563e134944d57a117c3a13858bd68 : SUCCESS in 10m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aee5881e7dd14df9bb2bb7772af1a494 : SUCCESS in 13m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/78f38d9b29144536870282394ec2e7ad : SUCCESS in 7m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c829eca7213e487bae97dba417cfebac : SUCCESS in 25m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a649c5440ab41b1bb6149616fca074e : SUCCESS in 12m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5be9e9da1af4f68a139a520f42727a5 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8682bbffe3274888acdf0ede785436ad : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/60335be3832f43e182efd84725c66661 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef462ec134654781b7206fb90c93eea4 : SUCCESS in 8m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/32dd46f44f13402eb8a1175423e7bf5d : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a9cc4c62a3c446a89a8e803ec12eeb0 : SUCCESS in 59m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/455294bb5acd4f489698bf357182e546 : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"eedbfa0eb0ae69eba00c09074d80a9441ec45a45","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 15:40:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23ee26eb269aead058acad4eba15c3e9d4019678","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 16:17:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8dbaaa157022ec9508526c22b32d7e7c861a5ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 17:03:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fe69afb511f1ce5a4bf4b45544a5603d6f15c43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:19:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b310423f33d046bc889600d23a13990b : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6e4c0cf2f149628b1e9f50d922a9d8 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f5213e696ec490b99d3de4071f59a2b : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92e7c3827a884fa182479e2ddf45479d : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98b634333b1b431a88912bf402b82b01 : SUCCESS in 6m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9deab2525ac4cad829d34abb87df1d3 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/139d2111f57b41eb84cd4a19c901bf05 : SUCCESS in 12m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/31cc7d36e6f24098b30e0c44523f8980 : SUCCESS in 11m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f18590c22940477da64f03152805e172 : SUCCESS in 12m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40718b7fa0534f439b214d188a779f0a : SUCCESS in 7m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/07ea049175a44fca93594781243848fa : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6803239dafaf42d193bea0e3e47a6beb : SUCCESS in 9m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b9ace16b42d240758906d60bf3f197ef : SUCCESS in 16m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/213fb09aa70c4bb68574135b7f3f3d24 : SUCCESS in 42m 23s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9334cd7d8b4a4595b0fc4d975780b328 : SUCCESS in 6m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8086434b7cc1442e9237e2df1ce77439 : SUCCESS in 38m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ba9c098717bd4f6ba5ae01cafcf714d0 : SUCCESS in 14m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c1c16104be94df9939c50308dd2a435 : SUCCESS in 7m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/110e9e9fb68c42d38b09b943762001d9 : SUCCESS in 21m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9bab5ed2c3054ef7b2accb7ee320afd6 : SUCCESS in 8m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3cac2e4d7e8a44ff976979de5407733f : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bc560bccf662452ead2d2f28336645e6 : SUCCESS in 18m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26cc5814ad53478893f3c2ad4aeeaaee : SUCCESS in 1h 03m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/143caff1e2164764bf15cade030dfd6f : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3792d2a0f744d9b3e62de8e266de415046a4c91a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:19:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e76ee52a6b97ec74294b58f39d24a15040276f3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:36:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3f82cc9bcf49fe878fa31e18fd3770 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65969d977223957c89fcc040aff42d2538dab622","revisions":{"65969d977223957c89fcc040aff42d2538dab622":{"kind":"REWORK","_number":1,"created":"2021-03-03 13:50:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/778466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/778466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/778466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/778466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/778466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/778466/1"}}},"commit":{"parents":[{"commit":"32ced57cbac8c2713f982bb094b9b2ab997ddea3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ced57cbac8c2713f982bb094b9b2ab997ddea3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 13:50:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 13:50:19.000000000","tz":0},"subject":"update constraint for oslo.context to new release 3.2.0","message":"update constraint for oslo.context to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4bca5c0ae21428d83dcd156d05bcd49b22f2f631\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Ib790097b6ce42611258c1ced57fcdf51183cd70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65969d977223957c89fcc040aff42d2538dab622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65969d977223957c89fcc040aff42d2538dab622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
