)]}'
{"id":"openstack%2Frequirements~778477","triplet_id":"openstack%2Frequirements~master~Ib76b130734491e9a026fb3cddf3c1889f2658f04","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib76b130734491e9a026fb3cddf3c1889f2658f04","subject":"update constraint for ovsdbapp to new release 1.9.0","status":"MERGED","created":"2021-03-03 14:45:04.000000000","updated":"2021-03-04 06:46:58.000000000","submitted":"2021-03-04 06:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778477-new-release","meta_rev_id":"ef4e93ae0e6e477348319ab38bffdb8372085b2b","_number":778477,"virtual_id_number":778477,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 06:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-04 01:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-03 16:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-04 01:47:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 15:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 16:18:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-03 20:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 20:46:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-03-04 01:47:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ac12e3b25e77946e7a96640328b6f9ccf46271d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 14:45:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d903b9efbd6d50ea9e56b1ffdcd6ba2dea0dd3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:32:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/007d3f018cfa4a418a45dbba92b98679 : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f6399d179081a25b870dbd23f09edbc8ca68577a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 16:18:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57af566679c623161b389d39aa4b3def8ad574bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:16:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e39e4b31bfd4af581f48d1c6fa31013 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22263b18f254f1b80b10978a98ef901 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c83cce1721e448148a72f1fbe28df9de : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6e676b6baad45748797870cc71f200b : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cd9428dc646440186226cbd32c4e2ad : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d198259b61884caea38d1abfb315927e : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88b155bded37484b98b884f0b0a8459a : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/367104ce7e4843d7a0058987da5589cf : SUCCESS in 11m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f1104b5b3eee496c8b134b78e7619d91 : SUCCESS in 14m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/02149ce288044f4e9342003f43a857c9 : SUCCESS in 12m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a711ea912cf49fe8d518f020dfe698c : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/084d1f760bae4fe9be0e9c7e2bbb0277 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4323ab6f153f4aa181a4c435d56604bc : SUCCESS in 15m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/619d00e73fcb4de6a78f40b42c53e77d : SUCCESS in 12m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6be4e3c9d7034d6785beb189f36e78cf : SUCCESS in 18m 34s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e958d203f93a4e648a10d45f34a065a2 : SUCCESS in 6m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/540819e83b9146efba942d21b651840d : SUCCESS in 35m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1df08b4f67474bd293598236df42b6b7 : SUCCESS in 16m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6a8d2bc52a4a4dcaa69fc32af0d3a447 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/255b3e084e6748ec9b824a351eb334a3 : SUCCESS in 19m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/48a5880d2f384a5eaeae967a3ad523b7 : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/493baa03900f4163b5954845dc321f23 : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea4388aea308480aa290b203ca4238b8 : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24ed7ff7ab034010bd6bec89b406302c : SUCCESS in 1h 02m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75bca88f3dca43deac3a6d84f96d8c1b : FAILURE in 1h 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f1063f8528453e50f5f8122026fcde2a1bcd649b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 20:46:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5eb6c42fde6463da0b91dda62f784fde1083ea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:26:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2453bf206fe34b3fac3e18292ed3cbce : FAILURE in 30m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a64706209fbd6229f42db040f5ab74b02b432e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:03:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6caf99da9c841b7b8fd3fee363711c9 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/228c98637510446ea463fb85d53691b0 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/577b7beda8b640b5840d32023bfea777 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6ba74d4184340d6bb993611dc036c96 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb7cdb47b8944a2999142d68897046c9 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24d08517fc564cf284f147dadc052d4b : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc63b4d5203d4491b56d98ae0e5c9538 : SUCCESS in 8m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e260e24d3113491ca892a84ffc647e4e : SUCCESS in 10m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ab14b0acf7040d6a26a8bc01421e794 : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/518fd24658784ac2aa9371e4cc3eca61 : SUCCESS in 12m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d5fd922a7ff45038764d8d657abaf03 : SUCCESS in 8m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53dc7caf1ed74f5b8e73b11c972eb70b : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/250cd00ebeb844c7b80d2600608c2ab8 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9fc689c585bd4d0f92c832b20506305b : SUCCESS in 11m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06570024a6ef47c18379f0b71d90f577 : TIMED_OUT in 1h 01m 11s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b89e0f090b1446299d50a8043c234807 : SUCCESS in 5m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/48dbe132fdca42f7803e01cdb1d21292 : SUCCESS in 30m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7fe6c2db0f5f4946a611cdd7a3e2fb10 : SUCCESS in 11m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/296b1322277f472c996126092dc6c1d1 : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27b383a1ff7f482aa61a024c020b2c1d : SUCCESS in 20m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e952ef2a33364233b60f39e0fa15bfa5 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f66cb4d30bd4d5b92be9729832d9f08 : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/734933ed894e403c887894a4c8c82371 : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81097b59e4af42469de37101eb1caf2e : SUCCESS in 1h 06m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2b6b8061cfa43dcb108e4e489c8e035 : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c67cf102572c0498f6e0273395a9d0eb25cc2ebb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 01:47:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"006dc7d08de528befdb6182fdf518f1420676a4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:24:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a4c12fe243bd42f5b7d360ebb205b54b : FAILURE in 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"57c7a91c229cbb18b3f212a70c5f9d1349cca06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 03:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0860de3c6ec94791bdeb75ab21190764 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40cdd7b642b44e23bcf26e618fc92792 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c45a70e4453d43649adb598e4ac9f536 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b8a418b13f5479ca5903ce5b194139d : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/893dddb81d5749e1b9cf36b9236ef0eb : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/780e18279afd4060beb80d019f7b214c : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4712ebb20e24389b17216d6f4f7518b : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/362eafef32d940799684b8e9b5a3e954 : SUCCESS in 10m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/12413aa59cda40a4913d580967ee6cba : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/53b0ac6832724dd69cac0d08d163a06d : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c535c7a93f1244f4b421eb0fec050f5b : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5ec735ad562d4c7dbe892e7b54890e76 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/648f2bbdaae64cf093bf54f72183be6a : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ddb4ff03e4ea4400a65c789080a47de6 : SUCCESS in 10m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2fb8f4dcfecb464e9544c1af369ce14d : SUCCESS in 13m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/44e6bc0b6c234a24955f5c614a69465f : SUCCESS in 5m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4bd7bf3d7ce0435ba9fdaca236d4addd : SUCCESS in 29m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5e720dc53b1e47cc8b52a65212a7c41b : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3cf3fa70d30408faf833b6826749b23 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc2c958ef45d44609bc671321223a95d : SUCCESS in 16m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b80c38688ff84de999d4e139f160c6f0 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aeeec5fef7ff4cb48cd368786e638c0b : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c26253e50a24373a4f18476390b77ef : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a31f9fb35faf4aa7a691665e068d339a : SUCCESS in 52m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da80e40381ca4aff9bea5ed7c04efd4e : SUCCESS in 1h 32m 07s","accounts_in_message":[],"_revision_number":1},{"id":"30213f5fb08a8027d13e48da38e790dca461c40f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 03:23:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01251c97fbc3399aa499ff5f8244bbf53a65db3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:45:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e56e32c88249bd9e0c07d96c7e9fa7 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2071b5b9d24344bbbd7d52eb2c3f62ee : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/550c5f9cca244ac6b15c19ace4a2c2bb : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed95380bd4cb43b0a530b8ae08f229d0 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abdccfd12d5a441faf857b845022e975 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c85227736ad941a796e2598c27b3003c : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe7544010d8f4529ab180fdfb296327a : SUCCESS in 7m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b0fa9239d164f97a547e86e80f786b0 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ee73addc4b82425b9975a6cbfc447e7c : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6327c3ef434e445ca3cca1b20cdaf575 : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a2a583d89904e79a8e34d1d0aefeb12 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58a42088e4f64a368483481e611f2319 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c6230b325994ef9949ec459815a2c85 : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ec4a1b2a2a1842fa843be19cd860d013 : SUCCESS in 18m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/09f4375785734468821c41b89c7d4e4f : SUCCESS in 4m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5b271e010e24a079cce391ed563de74 : SUCCESS in 38m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e01e90dcf0b413ca3889958edc8b11d : SUCCESS in 12m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99f6f5fa5d20482099b5c6e9bf2e2244 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23c5ebd328c84530afe25c20135d574a : SUCCESS in 21m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3027c30674eb48f2b8486a7fc41a94bc : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9cd2d9d5152446f09f298d7ea3b79c95 : SUCCESS in 6m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/554f7290b71d498f8847e606ffa3a900 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36af2e70b61441c9be7f6f28ea5bd933 : SUCCESS in 48m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f17a16cbc3543c8884a32b8cd3a84bd : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0f18dc8a87f8579f86be035ade3debf74b4b3bc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:45:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ef4e93ae0e6e477348319ab38bffdb8372085b2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:46:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf4d70a536749f593d942afdf78428c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fbb7eea70f66288b637932676ec6036ba819a69","revisions":{"9fbb7eea70f66288b637932676ec6036ba819a69":{"kind":"REWORK","_number":1,"created":"2021-03-03 14:45:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/778477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/778477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/778477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/778477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/778477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/778477/1"}}},"commit":{"parents":[{"commit":"32ced57cbac8c2713f982bb094b9b2ab997ddea3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ced57cbac8c2713f982bb094b9b2ab997ddea3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 14:44:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 14:44:55.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.9.0","message":"update constraint for ovsdbapp to new release 1.9.0\n\nmeta: version: 1.9.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: I2029bb7d2cfee207b4f71d881ce249066570560b\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Ib76b130734491e9a026fb3cddf3c1889f2658f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fbb7eea70f66288b637932676ec6036ba819a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fbb7eea70f66288b637932676ec6036ba819a69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
