)]}'
{"id":"openstack%2Frequirements~778495","triplet_id":"openstack%2Frequirements~master~Ib094583f783e12f264a7d7538ccc75f068d18ce3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib094583f783e12f264a7d7538ccc75f068d18ce3","subject":"update constraint for ironic-lib to new release 4.6.0","status":"MERGED","created":"2021-03-03 16:10:47.000000000","updated":"2021-03-04 14:57:53.000000000","submitted":"2021-03-04 01:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778495-new-release","meta_rev_id":"f990f40939419eb76cbe547d35c9a8ab86d7876b","_number":778495,"virtual_id_number":778495,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 01:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-03 23:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-03 16:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-03 23:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 16:20:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-03 16:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 21:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 23:17:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"44692bd9f0c0543cf5ed8ed7def0c5a86544c87a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 16:10:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43975effc8357ed549784c7b97c5170609dd3ea7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 16:20:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f97c669d6da6963b2175bebd4d9f063f7a2c211","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:53:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/71601da23c7d44ad8a69634995f414e7 : FAILURE in 36m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d75d6079965ae983f675e9454e0cb74559a43212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:13:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7323dfe6354b4c419439b1aab6053181 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd3fd71078443e89284c6e4a4a10da5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/140371010c1e41c8b6abb276fc30041f : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6089b131641c43f09982825dfe52eea4 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d98a7b98d03e4ecfb8fe72af542e4d61 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eec2340b6ac0450b806e93cb9ef54aa9 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1e3d59f722041a0ab30bf08ac270749 : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f8636c92e0fa4686be3321d40bae3fb4 : SUCCESS in 11m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb37ca24b38144e1a4e290a2fa5028bf : SUCCESS in 11m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ecd62bf281942fca5a8c783b5849b18 : SUCCESS in 12m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d3a6632e917482aa2d1b108c8a3a920 : SUCCESS in 8m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c11c583e9119492a9a62ee2350cd8946 : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e63fc1f3ceb44008add7f2c83dff94f9 : SUCCESS in 11m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a67104f708943ecabfaaf909a09c230 : SUCCESS in 13m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d2bd8828faaf44f6894843fd67a4a314 : SUCCESS in 19m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e1fba967b5fa4688beaf9845a6e61b0e : SUCCESS in 4m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f15a7db3332d43f3a9ca34c879cd439c : SUCCESS in 29m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3d4c06baf0984e3b8f5c930189bed3a1 : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81293f97114c4c02aac04a1af9437d3c : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/418d61acc7544b6c9144300588705121 : SUCCESS in 18m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/518307fbbe6f4aeeb016dba568c8a8bc : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b83679ae48a34f0db2ad65f8bc53dbbe : SUCCESS in 9m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a4eb17586fe4e0f96d1b9346a181709 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6effe0e1f902486babf41c9824c9e350 : SUCCESS in 56m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b39c69e1095340149568571811a3250a : SUCCESS in 1h 44m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1099b2b5615d7d7cd578d5ffbf91b3d3ea2e12c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 23:17:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d4ae63b0a856fbacd54f30c85514a4a60a24b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 23:18:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cd28be8484adf88e9a076a1b9b49dfde795e20b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:36:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc3dfa918bf459a98a44520389800c3 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28c8d37056f41c6969a91647a11c2d9 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81ce9b61d290414d9fea062bd063d60a : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5cf8ff9289442e7a157f688464d1edd : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d01feb380dd64cd0932579f35e91a776 : SUCCESS in 6m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8f5e16e3af045dbb561adbe36674ee0 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d3c34e8c742a4eb0b3cfef5124d3ea66 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa0005500d67498ba9443997efd08156 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b846fed2bd564dc786c9921c0aa2fb31 : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/20ebb329545a4cadba37d84bcc0bbe99 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/43a369fb25504c02a0f19f746dfd4171 : SUCCESS in 12m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f515e0b9294545089406a69cb1dd1fc9 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f6650f832da4a128f79c7c7a781e1d3 : SUCCESS in 13m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/865e9b68be184ccb8074db04dc864a5e : SUCCESS in 15m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/af922e3679904ac192df7ba56ead2799 : SUCCESS in 6m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/81803f59e04c42d2884258383e2698f9 : SUCCESS in 37m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e4229842d4f246a99a1747e507b641b3 : SUCCESS in 13m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b491eb76ce44522b842427bc2e7c8cc : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/785c652b86d64a33b2fb6d3b70c2932d : SUCCESS in 19m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e72309740b504b828fc678478543ece8 : SUCCESS in 11m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ede5cfe08eb41df843e5e5b8a7113c7 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8881316e352b4e8e8424681c39525fc0 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/651fc5b1a16e491cbcf70c1c78338c8d : SUCCESS in 54m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8da1767bc709480c98cbaf7df5fb5b7e : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"02c9cfa3a514ecf6b5b807d489c6cf5ef99b591a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:36:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f990f40939419eb76cbe547d35c9a8ab86d7876b","tag":"autogenerated:gerrit:revert","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-03-04 14:57:53.000000000","message":"Created a revert of this change as I7ead4753427d3492d02fd8256768e6efba52a12a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"347e83cf56faa4c634cfa54e07c9731585c2e13e","revisions":{"347e83cf56faa4c634cfa54e07c9731585c2e13e":{"kind":"REWORK","_number":1,"created":"2021-03-03 16:10:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/778495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/778495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/778495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/778495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/778495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/778495/1"}}},"commit":{"parents":[{"commit":"32ced57cbac8c2713f982bb094b9b2ab997ddea3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ced57cbac8c2713f982bb094b9b2ab997ddea3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 16:10:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 16:10:44.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.6.0","message":"update constraint for ironic-lib to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iac5aa5de01a36d32dfa41a8a821874442f6984ca\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Ib094583f783e12f264a7d7538ccc75f068d18ce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/347e83cf56faa4c634cfa54e07c9731585c2e13e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/347e83cf56faa4c634cfa54e07c9731585c2e13e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
