)]}'
{"id":"openstack%2Frequirements~778496","triplet_id":"openstack%2Frequirements~master~If7f8c25d77c8c290d4da6c8e447416245642e930","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If7f8c25d77c8c290d4da6c8e447416245642e930","subject":"update constraint for metalsmith to new release 1.4.2","status":"MERGED","created":"2021-03-03 16:10:51.000000000","updated":"2021-03-04 01:42:15.000000000","submitted":"2021-03-04 01:36:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778496-new-release","meta_rev_id":"b3b58dea65482f5cb154cbc12e88f3be7af70f5a","_number":778496,"virtual_id_number":778496,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 01:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-03 23:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-03 16:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-03 23:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 16:20:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-03 16:45:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 21:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 23:18:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"72c1933f84d6830fceb904efe3f3c5dcd8bffe5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-03 16:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffb65e5750a62e79843df4848437eaf552810c34","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-03 16:20:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91a8e82e4edb1388c8048adf1b46ca6d4fef9764","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:45:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cf42f655e52f483794ad69369cc06b91 : FAILURE in 28m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b91c5c3fdc4333ff0627114b12ff609b233fed36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:09:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751eb6eb094c47468310cef2401752ab : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684db57cc5104c609f9ba4e23995de5d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed3158ae749e4e9ab2656bf7fe724c6e : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23e73482fdeb4314806dd418dccb270b : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9eb3b7924f094f679e8666bf4c2bd535 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca612994e8444820bff9c8ab0b70e560 : SUCCESS in 9m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb1879af19084e2ea31dcdd9b87b8caf : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc0fe794f7894b6b8eb69db122655021 : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d5e6d65c9eb5463bb4b30a95c09b7179 : SUCCESS in 11m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8994d464ddd94da8af8d968cb52038dc : SUCCESS in 10m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5b8c9f075da4f0b8633e3e4c0f5b432 : SUCCESS in 7m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73addbe4d9d94fb1885f77db0766b01b : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/421e8487b17642cb804ae1974378278f : SUCCESS in 11m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/498e59105015499ea142d51e15d0cb3b : SUCCESS in 16m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a450521327514f0ab86868712cfe886e : SUCCESS in 14m 31s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f158e42877784dc7be12252fb357460d : SUCCESS in 5m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2010ad57561941ce84a0034e2afa0721 : SUCCESS in 26m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37d3fd97cb514bc9a3b7c1a57f2bfe39 : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac6948f7d6e748e09dde5d3096f2d196 : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7170b7a890ce40e5b4a651bd07246edc : SUCCESS in 20m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9259e68abbf451dafc4139c44304328 : SUCCESS in 8m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/beca8faabe37440f9771c5f0ff6a2a14 : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/354bcca548ba4c8aa0498042195d38ee : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b3c5092beed4527b9cfd785a331bfd9 : SUCCESS in 1h 07m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63b354047b64429997ced91e7b86b3bb : SUCCESS in 1h 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b6160c9593cf77330899fce81ed550c4f924fd33","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-03 23:18:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa7532d15613a0f2e240bf2cd678cfed2b5a416d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 23:18:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9bc0f03a1ec388aa5019d29c9f216ac37029b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:36:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b188f3ae5904c2299ca2d42adf5f041 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f704fef868294a009a2c3539b3f2c6bb : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1baf7f38ca9f4bc0a8d4553b37822195 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4add96d3a79d42b6b9334f4447d8ba9d : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72f79a7488184bd1be9087e9e8f4eacf : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6ef28697bd04202ae32ada6aaca171c : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b26e11369a24b06a3dc1f397dd09331 : SUCCESS in 11m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b1f408a691b4add8f9195ce91807414 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f637c99a01084a44a616a99c45bc7cf1 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8a32d1b444d4490a8a50de80265fb6cf : SUCCESS in 7m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7d1914a91a6b4c3d95296cc4148cd993 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e782417c08bd448c8400774264c79c06 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ce394f62764f4624811a03e3bbc4301d : SUCCESS in 11m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/309ea283c09f46eeaa2370976b602285 : SUCCESS in 13m 41s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3b144f58c4a143968dab1151b307a65a : SUCCESS in 5m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7922fe63c5204d4aa9650c2035898ceb : SUCCESS in 26m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7933e3f434c3495392d2e0c60dcf2194 : SUCCESS in 12m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/592f78b8b6d248c78a45c1cd37f67555 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6aad0b44dfdd41ec8cc1d0ba9d3e88ca : SUCCESS in 20m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/46848db4b42148c7a444ba443d5f3316 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0aa0c37475645b880debc51095d676e : SUCCESS in 8m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/99540785d7c248d8b7b2654484330b45 : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb7301492f0c44a49667d778849119b1 : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/233507fdb61347c2a3e438c3816116e1 : SUCCESS in 1h 27m 53s","accounts_in_message":[],"_revision_number":1},{"id":"982a6d60074287f7ba8535a29e3710ded3a79625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:36:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3b58dea65482f5cb154cbc12e88f3be7af70f5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:42:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76be87ec669345ff877c2e50758bd2e0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"214f514b565026dfc682fc1fb6a5e79abf8a9404","revisions":{"214f514b565026dfc682fc1fb6a5e79abf8a9404":{"kind":"REWORK","_number":1,"created":"2021-03-03 16:10:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/778496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/778496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/778496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/778496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/778496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/778496/1"}}},"commit":{"parents":[{"commit":"32ced57cbac8c2713f982bb094b9b2ab997ddea3","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32ced57cbac8c2713f982bb094b9b2ab997ddea3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 16:10:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-03 16:10:47.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.4.2","message":"update constraint for metalsmith to new release 1.4.2\n\nmeta: version: 1.4.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Ief045a21c59f4a11743c9d870eb49caafa10528d\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: If7f8c25d77c8c290d4da6c8e447416245642e930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/214f514b565026dfc682fc1fb6a5e79abf8a9404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/214f514b565026dfc682fc1fb6a5e79abf8a9404"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
