)]}'
{"id":"openstack%2Frequirements~778614","triplet_id":"openstack%2Frequirements~master~Ia2899b52214fd4babebbbe4fd84754611b8b6a23","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ia2899b52214fd4babebbbe4fd84754611b8b6a23","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-03-04 06:20:19.000000000","updated":"2021-03-06 19:39:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15a15b19b6cd2110c55e591c38bc02d03086dac8","_number":778614,"virtual_id_number":778614,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-04 09:13:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 06:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 06:55:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-04 09:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04007468b2094303c3a12fd4fbdfff8b78f9cfa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-04 06:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"997572333c43c3af766810b4a8a8c3cd75726b27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 06:39:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c866bfe6e114ab1bcc5aba99d04e20a : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2c7e750345c8c0c09922ee695c50a4f397e8b1ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 06:55:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e87648c78fe7289be300a1f8e242d99166139ec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 07:01:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"048f6a47b62ecfeef04dfdcb73fff9190a02356b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 07:02:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df8e0e8d69e97f32b8dbdd20404fbbf9c299cc81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 07:34:47.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/885db80bd1fb40628a3ea4cc7066a1a5 : FAILURE in 28m 19s","accounts_in_message":[],"_revision_number":4},{"id":"44f55666283d13a4e293588e7c832ff3b3bd2455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 09:13:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abaf4c49afc94967a7b3af5721bda6aa : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba41e3de327d4ffbb80efca80b637dcf : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c5c5a782c6249c692b44d1ee308ebc2 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/863808d126514154b1424989287560bf : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/622e2e3b76e5487991540abb94a068e6 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c825b0a10e440f9bb26a175673e59d2 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61d03048b7224144964b24979fb21695 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1eef33228396440aa1d105d36f284dc1 : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c7f2c4165c4c45319a888d2296f0aa49 : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa62bae6687a42f485bbf9f79d32918b : SUCCESS in 7m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fb9be8cd595249d98e05ed18b1f33548 : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d3d77f63d0cd44a6ae9aa870c0efb7a5 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85709df3adfc40a6bd096af800596bcb : SUCCESS in 6m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ccdcf3de8ce64a7183f0c9f984a2cc5b : SUCCESS in 9m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b3292bce1e54ad9995e9b2e1d42eef8 : SUCCESS in 14m 20s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7b49cb7f47f345fca56f821a12d55ccd : SUCCESS in 4m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1249036c1d2b4b3c9553568a5aa852ed : SUCCESS in 26m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/193110b4e1894f6394eaa0e12dba5b30 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b154f6059364ce8873ed37f6997e130 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32c24d63f4e64ae4ab9ba9843fee1ee0 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3a725ce956424dd083a2eb03a2e567fd : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae6f7779e0bd468b8d56a71cdd9ec811 : SUCCESS in 6m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/550697e3994b4487a7cde039eec463f4 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c14f1fd902e41d89a3f3ccc9348504a : FAILURE in 1h 09m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9eccc4b42abc45c5a489d5a1434c3007 : TIMED_OUT in 2h 05m 13s","accounts_in_message":[],"_revision_number":4},{"id":"67cd839ccff21782ec5109d7f49ced2e226d63d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 16:36:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"195cb1887c439d4b006eaaff70c334fbadb2c21d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:28:27.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77b131d7d2bd409e9e3c9323dc1779f4 : FAILURE in 34m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6b66d24f976dca9295b0d29e6246a2b8b9b0724d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:31:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21b97d19d044c81b21be1c177de4850 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28c3216272647ad95129b7815098120 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fc2636aca7d42bba5bc1856421e03f6 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99aa33a5163b48ad86a0c741cc94cb8b : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07453dd709fd4e539af792ab0c9379e9 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eda3fc447e2547eaa0a664bcc513cfe7 : SUCCESS in 8m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fed23363056944efbb8694644ae7a170 : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2640b6265fd945a9be9a16037050365a : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a683e09fc9b249569ed7b795333c38e0 : SUCCESS in 10m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fee9f7e0772e49f2a1fb2931adbe6642 : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59ff4aa033d3455f863f5734cd12f61f : SUCCESS in 8m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a8bf7048e7343c2a8beb7f47c0ac823 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/027986417d0a4242ab5dc7c29e3b7167 : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9cfa204f25bf44368e6e253c00344476 : SUCCESS in 13m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/028844c65e6e461c96277159a2e0cd79 : SUCCESS in 22m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2353109e85654f12990b9a1bda55b14d : SUCCESS in 5m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/24d301896cb84b79849a8fa0ba92b1b8 : SUCCESS in 36m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/847d3d18e93d481b9be3f63a2c2a9240 : SUCCESS in 14m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e6e3cc104944a7dbd497660cd023278 : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a24d014f7374846b640c5b16fa07ff4 : SUCCESS in 20m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a5fa77198f84fee8cedee59a3682969 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/243c86f76d8d4b648d184d114a4f6dcf : SUCCESS in 7m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c0a3d2f2e6349219d6f0a5de1bb4f85 : SUCCESS in 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30c8f946d3004535a44d6c4da925728b : FAILURE in 1h 10m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e171f2f667b64b6a9a6d51346f370e91 : TIMED_OUT in 2h 05m 01s","accounts_in_message":[],"_revision_number":4},{"id":"69052cb71c0163ae23c1cacfe664222b77c7f218","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-05 02:02:39.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"c411d79c033b7185e9383459d3998df1c411cdcc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-05 02:05:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4edec65e775aae539c132c4db0e33ce104849c7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 02:43:48.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb2c01e86b044f96a043ad88be2c7961 : FAILURE in 34m 25s","accounts_in_message":[],"_revision_number":4},{"id":"15cd829439b6d9b11dd4eeb28a1a9c28a7e88a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 04:13:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b1817a822943fda5ef4453d08593bc : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc2ea41bc9df456a9887028e18533117 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faeec40973f147e7bf40936563557148 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c712df1caa3c4ada9b5ed44b24d4a1f1 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c72879247335444d8bf2df33c42a6481 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/029f1abc08de4323878320e61a8a767e : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c8b9a4127704cd783e923e830f707de : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6f78a4f93004eb497232bb883c0dbc8 : SUCCESS in 17m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/336698000f0f46e79328afede8f300ba : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6a186a8209c846298f8e9a4a50075906 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49b4fa8c3b444571aedbdcc5123e8331 : SUCCESS in 5m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5fb2293baf474cbdb3531fb9fda39f87 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7af525a594bb4a6fa0736bbae931d1fa : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ec7e1177215245829732070b8fba2d64 : SUCCESS in 10m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fcc94c0dee4846fd8ce8049dca0bf5ed : SUCCESS in 14m 44s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/67472606886d4fcf9adf8027b2e7f97d : SUCCESS in 5m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1bfde4fc379d49f99a9439fb127a6e26 : SUCCESS in 31m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/39a9ed3b7dee4f8fb2f9a79af86033ea : SUCCESS in 12m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4357f15d9c4a424db772031427ea46bc : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/611cb74918fd44c48ef1cd297159f16f : SUCCESS in 18m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1ea814c56eff4352a6c5e0f079d32476 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e0bdaba72f04210921946885dfd59ab : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/08656d9e3af3487dbbc738961d981975 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/468c45fcb1924a6dad695f5ae08c4ced : FAILURE in 1h 00m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e63bf59ea684656b18af6a8fd4e8c77 : TIMED_OUT in 2h 04m 03s","accounts_in_message":[],"_revision_number":4},{"id":"15a15b19b6cd2110c55e591c38bc02d03086dac8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-06 19:39:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d559668f8b2d2e3ee67675f20c8857eadc3a50a0","revisions":{"18bd5202c04da8c24a21b737b305a600fd33a2ca":{"kind":"REWORK","_number":1,"created":"2021-03-04 06:20:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/778614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/778614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/778614/1"}}},"commit":{"parents":[{"commit":"d4674b401e42590dd6882e7850e4dea828cc67a9","subject":"Merge \"update constraint for neutron-lib to new release 2.9.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4674b401e42590dd6882e7850e4dea828cc67a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 06:20:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 06:20:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia2899b52214fd4babebbbe4fd84754611b8b6a23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18bd5202c04da8c24a21b737b305a600fd33a2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18bd5202c04da8c24a21b737b305a600fd33a2ca"}]},"branch":"refs/heads/master"},"4a64c59d52a2018385646b2890ed19bc97b0faa6":{"kind":"REWORK","_number":2,"created":"2021-03-04 06:55:11.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/14/778614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/778614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/778614/2"}}},"commit":{"parents":[{"commit":"d4674b401e42590dd6882e7850e4dea828cc67a9","subject":"Merge \"update constraint for neutron-lib to new release 2.9.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4674b401e42590dd6882e7850e4dea828cc67a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 06:20:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-04 06:55:02.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia2899b52214fd4babebbbe4fd84754611b8b6a23\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a64c59d52a2018385646b2890ed19bc97b0faa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a64c59d52a2018385646b2890ed19bc97b0faa6"}]},"branch":"refs/heads/master"},"b374c8f2e63bff62254d784473b827744228f1ae":{"kind":"REWORK","_number":3,"created":"2021-03-04 07:01:37.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/14/778614/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/778614/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/778614/3"}}},"commit":{"parents":[{"commit":"d4674b401e42590dd6882e7850e4dea828cc67a9","subject":"Merge \"update constraint for neutron-lib to new release 2.9.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4674b401e42590dd6882e7850e4dea828cc67a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 06:20:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-04 07:01:33.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia2899b52214fd4babebbbe4fd84754611b8b6a23\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b374c8f2e63bff62254d784473b827744228f1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b374c8f2e63bff62254d784473b827744228f1ae"}]},"branch":"refs/heads/master"},"d559668f8b2d2e3ee67675f20c8857eadc3a50a0":{"kind":"REWORK","_number":4,"created":"2021-03-04 07:02:05.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/14/778614/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/778614/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/778614/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/778614/4"}}},"commit":{"parents":[{"commit":"d4674b401e42590dd6882e7850e4dea828cc67a9","subject":"Merge \"update constraint for neutron-lib to new release 2.9.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4674b401e42590dd6882e7850e4dea828cc67a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 06:20:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-04 07:02:00.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia2899b52214fd4babebbbe4fd84754611b8b6a23\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d559668f8b2d2e3ee67675f20c8857eadc3a50a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d559668f8b2d2e3ee67675f20c8857eadc3a50a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
