)]}'
{"id":"openstack%2Frequirements~778691","triplet_id":"openstack%2Frequirements~master~I7ead4753427d3492d02fd8256768e6efba52a12a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7ead4753427d3492d02fd8256768e6efba52a12a","subject":"Revert \"update constraint for ironic-lib to new release 4.6.0\"","status":"MERGED","created":"2021-03-04 14:57:53.000000000","updated":"2021-03-05 01:57:18.000000000","submitted":"2021-03-05 01:56:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":778495,"submission_id":"778691-new-release","meta_rev_id":"d791110f8b8fc9c5b933d2cf395293c2182c93a0","_number":778691,"virtual_id_number":778691,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 01:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-05 00:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-05 00:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"reviewers":{"REVIEWER":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 14:57:53.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-04 14:57:53.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 14:57:53.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-04 14:57:53.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1abceb4f272af2aefdcce7ebe4b928189b0e8591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-03-04 14:57:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e4914c859de3f7eb74dd03f33cb07226d784cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-03-04 15:01:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cbd8a5d0c682d470755677f7d5042df5874561a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:12:12.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a8113b4095944a4db15fafbc8740ec18 : FAILURE in 43m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ac9db3d9180d6e77b92a769b2497941e150239d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:59:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879722746cc84b50bfb26804d77a72f7 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94e675ba87e4fee89e85f8953fec0a8 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75bd0713c2546a0a9c661c14a38f71a : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b385a697cc734ff3bb472ac81709ba6e : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be29407f61814cc7833565282f6d9289 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdf37799ea174d7a952d5a4699bcd1e2 : SUCCESS in 6m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/980d7218db354a94ac031eae1ab3317c : SUCCESS in 6m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a0d108391eb41289de67a3a8c07fbbc : SUCCESS in 12m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14b090d0f71c4e3db20de18a01b3f5a8 : SUCCESS in 14m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c4f05e89fc8044ea9d3ad308002d8a0b : SUCCESS in 14m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb8b379ea2ef474ca5f46afb1baffa31 : SUCCESS in 9m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c04e1fbfed5446bca1d7992f6a315de0 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b143c179d66842b4a9ed28b2767d5e67 : SUCCESS in 10m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a9258d4f3074d7caf6ca577e4d075fa : SUCCESS in 15m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d75b17961e0c4427abe338949264e8f8 : SUCCESS in 26m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1d9eaa32c249499eac952f7479d174ab : SUCCESS in 11m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0cd2ecf1dfba462faa85ee35b3b401b2 : SUCCESS in 39m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f1aff907ae8403392fffdb90e7c02b7 : SUCCESS in 22m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/006c1ae4319545b88b95ec974ead0310 : SUCCESS in 7m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e216680d75e449ef979472a7a3f97fec : SUCCESS in 24m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/409e7f4ab35c471fa6167912ef856959 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/689fbf28f8e9494f84d7f1321d5778f9 : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e8ab7ddbddc4a4d8bb4f0e79681c93a : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd8b388d224a4cb88b8ae6c02b8b4f87 : SUCCESS in 59m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9286f94ca22544f3acd126bad698914f : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8fa2a471b82fbad98301ffd5fedadf34cf4f8e3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-05 00:45:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c19440f3c5df497f337bc609af05bad37303f7ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 00:46:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f61793e3ffeb5a99198a9200df3b8769d39090fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 01:56:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553c53e203824a92a3d816dcb54a7d88 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55c1c63d8c24d1ca8eb2875f24c51b1 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e4aadd2286476299aede50b01c1b55 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/711ec67c24154b699f5e8e2a32d46135 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e07e6936bcef4cf7b39406b6131391a5 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7638027576504cc39d076c7f5b74efcc : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8738e3f821d545fbb2680e947b4a1a79 : SUCCESS in 10m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46f96142067b453f95c2ed3ad426f9d9 : SUCCESS in 7m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7caa0a1a55784e809998b3c6829362e2 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c172cf9657eb4670bb672214f475f0a9 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/efe3c62e889846f4934dccdbec8ef3c0 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f6da3bdda6540beb440f4936382c930 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91378209db3541b9a81783ac00f96e4b : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e5c4eb56d3124e1ab77a6b7fd7b2fd7e : SUCCESS in 12m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a8214225819f4ff7af4ad7760cd04bdb : SUCCESS in 4m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7b2232dd31049be82b36ed9952a4e29 : SUCCESS in 20m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bbfe506b3dbc4479991a6891443ad662 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd2894b3baec47388e3b3c5f5547e622 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ca61d7da64b4fbd9673741678d0bbe4 : SUCCESS in 13m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c054d472530a4a84a750ad7fbf34d762 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/309c81ab3ede463c95556202747c443b : SUCCESS in 6m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8baff18a387f41c2b3cf3e4324e3a9d8 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e628fa36cf13426c936e023c33723034 : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9ff4851f2ae46a1b76913e8e38e8ec9 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b40551f765a5ec25b694a380f51dedc8a886d4d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 01:56:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d791110f8b8fc9c5b933d2cf395293c2182c93a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 01:57:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37e0675e2244df7a1048cd56a59aff5 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a907762cfdee24d78b0e2cf23857cdde5e7bc98","revisions":{"de102c713fa31955b6a19df8f81320f27001468c":{"kind":"REWORK","_number":1,"created":"2021-03-04 14:57:53.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/778691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/778691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/778691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/778691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/778691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/778691/1"}}},"commit":{"parents":[{"commit":"347e83cf56faa4c634cfa54e07c9731585c2e13e","subject":"update constraint for ironic-lib to new release 4.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/347e83cf56faa4c634cfa54e07c9731585c2e13e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-03-04 14:57:53.000000000","tz":0},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-03-04 14:57:53.000000000","tz":0},"subject":"Revert \"update constraint for ironic-lib to new release 4.6.0\"","message":"Revert \"update constraint for ironic-lib to new release 4.6.0\"\n\nThis reverts commit 347e83cf56faa4c634cfa54e07c9731585c2e13e.\n\nReason for revert: broke the ironic CI, see this patch:\nhttps://review.opendev.org/c/openstack/ironic-lib/+/778714\n\nChange-Id: I7ead4753427d3492d02fd8256768e6efba52a12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de102c713fa31955b6a19df8f81320f27001468c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de102c713fa31955b6a19df8f81320f27001468c"}]},"branch":"refs/heads/master"},"2a907762cfdee24d78b0e2cf23857cdde5e7bc98":{"kind":"REWORK","_number":2,"created":"2021-03-04 15:01:42.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/778691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/778691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/778691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/778691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/778691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/778691/2"}}},"commit":{"parents":[{"commit":"347e83cf56faa4c634cfa54e07c9731585c2e13e","subject":"update constraint for ironic-lib to new release 4.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/347e83cf56faa4c634cfa54e07c9731585c2e13e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-03-04 14:57:53.000000000","tz":0},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-03-04 15:01:31.000000000","tz":0},"subject":"Revert \"update constraint for ironic-lib to new release 4.6.0\"","message":"Revert \"update constraint for ironic-lib to new release 4.6.0\"\n\nThis reverts commit 347e83cf56faa4c634cfa54e07c9731585c2e13e.\n\nReason for revert: broke the ironic CI, see this patch:\nhttps://review.opendev.org/c/openstack/ironic-lib/+/778714\n\nChange-Id: I7ead4753427d3492d02fd8256768e6efba52a12a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a907762cfdee24d78b0e2cf23857cdde5e7bc98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a907762cfdee24d78b0e2cf23857cdde5e7bc98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
