)]}'
{"id":"openstack%2Frequirements~778745","triplet_id":"openstack%2Frequirements~master~Ia365959b8aa4777228684a9836bdbad7c95b2151","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia365959b8aa4777228684a9836bdbad7c95b2151","subject":"update constraint for octavia-lib to new release 2.3.1","status":"MERGED","created":"2021-03-04 16:29:36.000000000","updated":"2021-03-04 21:31:37.000000000","submitted":"2021-03-04 21:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778745-new-release","meta_rev_id":"fb01ae1a8fd608b4fdc4a34094546fe020719df4","_number":778745,"virtual_id_number":778745,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 21:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-04 19:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-04 16:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-04 19:44:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 16:34:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-04 17:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 19:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 19:44:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d3e5d1277dec64538152956df0d7336948a32c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-04 16:29:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53831bf553fa19d04ee4a430cfb1c77bedd1bcba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-04 16:34:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dcf68e8be7264358ddc7c494788cb7b761f50ce2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:21:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/255d0618cf794809895f95fa70e48ec7 : FAILURE in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b81d05aa565f8ff83b5e47c835210960d6bc9a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 19:28:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92c99f3a8cf483195be478641619b12 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c28b0122fb94eb49520f38df16b8035 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4f2b4873e9f41d98ec64fa849e4ecdb : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/469d6c9d874542f0a7ec610af23bb247 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01e13a87dc3a4f1ba9812309f2707459 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfa0d99f73a54ddaadb9c4f144c96645 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a16bd39f8d6143059f8e34574268704b : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/05df91ad0b294c34a95ac5c9a271061d : SUCCESS in 12m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4165e0775ae34c5890e475cc66cbfb8f : SUCCESS in 14m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9769b6dde302473981bc0dc2d6a49069 : SUCCESS in 10m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/19021e38990947c8a539665802dfffea : SUCCESS in 7m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e0494aea148a46a885ee7a33391e14df : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48dbef6b3a31427d800dd89c28c7c7af : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/02101f27c9274aed8776c1d440641ee2 : SUCCESS in 11m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e835ecb07846482b90279014d39aab39 : SUCCESS in 18m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fc71fecbf9274dbbbdfe67bd8cd35020 : SUCCESS in 5m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a478d9baa4d4f56b15fd7777a56aff5 : SUCCESS in 30m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/525c0e70452e44959cdf59e082a4158f : SUCCESS in 12m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c63de9e63c1e4f77a695dd8772fce210 : SUCCESS in 6m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a73e19ff16f4cbc9c79dc88104f9c01 : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/79efd3ec628448eabe6854d310b3e186 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9eed56546ec54bfe84b18df1bf4a1ee9 : SUCCESS in 8m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/948540b8eb804c1eaf25d79634d480e4 : SUCCESS in 15m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd3081a880b64951ba2886207478443c : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71d5a3d007d9404ea2fbd12738061e43 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":1},{"id":"59e30c270e900a8aaba02cccb3648cba45f81a57","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-04 19:44:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea871c8a1d90dbfac29282f37fc0936a94da220b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 19:45:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"612eaf292af0f0d3fb71dc80a06fae2265ee36b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:30:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282e3e082f894a1bbbc7bfaac0a8de3c : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a36c810d38441dba5b3ca7452fcb657 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b513aaadcea84a88be297c95b54f7bb6 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e9f250cd5eb42618c2ec90fd2c553c7 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/067f9b757f924fa4942f1a6a8b8c2aa7 : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec4691ab7e7a44bcb47c69d7bd35b9c9 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d901346c1314e13a4039a943d83c409 : SUCCESS in 8m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e192bbf81daf49cb85a7af6b546d6497 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93843442f85c42239e19059d07036ca2 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f9a41c5ffc5a481cb30177853649bd31 : SUCCESS in 7m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/39b5c19e58a14d5593695743820862e8 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97a26c5c00ad41ea9bdf121072c9c727 : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4f33be9e7e243109c3dfd47a8b669cd : SUCCESS in 11m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0c77938fcb284533b9fa16f351bf82e9 : SUCCESS in 11m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/15dc44b65f874719b5c993827ae5bf74 : SUCCESS in 5m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/092c4ea9676041aaaa012e59cd2c5c17 : SUCCESS in 37m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9d0b031bc7a8471b83fc57da37822b3c : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eab66b1f68c546dea5bf424916dbe669 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1318c37c3f094a42b7b98ebf04990f5e : SUCCESS in 18m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ef9e4b5aca1c452b82860bc9cbb32438 : SUCCESS in 7m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5cc4f81ed74415881a0fbdb44cc4e5d : SUCCESS in 7m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7de8fbedd1d43889955fa6eb80711f0 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74cac2bf3f1a4cd48f4c5538b6597e29 : SUCCESS in 59m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a75c399aba3546a2b2744ecf5707d13f : SUCCESS in 1h 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6868cfc2f54f447dc32f7cd2a7f92329ed37d49e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:30:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb01ae1a8fd608b4fdc4a34094546fe020719df4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:31:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68842c59e6c542c5ba85e0db036dbc6c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"669190a66739074f947d8b52386d2853f7c0e723","revisions":{"669190a66739074f947d8b52386d2853f7c0e723":{"kind":"REWORK","_number":1,"created":"2021-03-04 16:29:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/778745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/778745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/778745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/778745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/778745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/778745/1"}}},"commit":{"parents":[{"commit":"0a36eb2b2e6a811a7728516998fc0000310836f3","subject":"Merge \"update constraint for castellan to new release 3.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a36eb2b2e6a811a7728516998fc0000310836f3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 16:29:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-04 16:29:33.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 2.3.1","message":"update constraint for octavia-lib to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: Ice2a6d2cc5285169fb03d94201feb8205173851d\nmeta: release:Code-Review+1: Carlos Gonçalves \u003ccgoncalves@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia365959b8aa4777228684a9836bdbad7c95b2151\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/669190a66739074f947d8b52386d2853f7c0e723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/669190a66739074f947d8b52386d2853f7c0e723"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
