)]}'
{"id":"openstack%2Frequirements~778873","triplet_id":"openstack%2Frequirements~master~Ia4900fb8b37d3711c080e5d0c22b963495e2f852","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia4900fb8b37d3711c080e5d0c22b963495e2f852","subject":"update constraint for tosca-parser to new release 2.3.0","status":"MERGED","created":"2021-03-05 10:07:09.000000000","updated":"2021-03-06 14:34:05.000000000","submitted":"2021-03-06 14:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778873-new-release","meta_rev_id":"32c354ef7e544f84f19817c8b05c6435083ed3cf","_number":778873,"virtual_id_number":778873,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-06 14:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-05 20:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-05 10:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-05 20:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 10:13:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-05 10:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 11:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 13:55:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-03-05 20:22:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7704348da553d849c49e8239ae4eb4bae932a432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-05 10:07:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e9c1717ccddf909efbbad4cabbcf551bfae5b5e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-05 10:13:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b479e0e4b77efc20df2d0a4f5df5bb1f4cb9786","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:54:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a86f5a1b6a7641d8a0c87f4d0d90dbaf : FAILURE in 37m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a1650524f320113d9cd8ae2d8686d4c890db5f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:37:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1efa7a747c460e820bab14379d3dfc : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8107ea6a06947a2bcba2c93083197f2 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36d0a8adc0e44c9a9c9dbeba9bb2d31c : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85856c34ae24055b1cce1022dfbf2cb : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b3b1aeb35db43d4972218af0125a5e1 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0b9ad8ec88c4b60a11f7fc84b7376f2 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a28edd65366948a6b90e29a496dd5010 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cfc5627003284cf2867f5c89cdab52f8 : SUCCESS in 13m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79c250279c5744e48de9e1c28e4f9eb5 : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2735b6988c3e4142a343c847f3fc40fd : SUCCESS in 10m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/403bc42c91324a8ca4dd26e028e97294 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6907cd12f346494f893bed6fb6baa000 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6d199fc31624a38891582b7754c7491 : SUCCESS in 7m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a1a9a49136404a85bafff70a64b272b3 : SUCCESS in 10m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d24d4c6250341e4a582144bd7f9d710 : SUCCESS in 14m 03s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3f14da94a3ba42308237134db6f0f493 : SUCCESS in 5m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6542c23aff2a48429e4b4aae9acad954 : TIMED_OUT in 41m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d2139b88e1544e64b5a83b585b34f619 : SUCCESS in 13m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63737e01c2da4efeadd93b40bc78879c : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da383be02a034d178e7b55d60fb27446 : SUCCESS in 17m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4d0df242184e48dc922ecfac09172be4 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbf728b0ecf64e1298000de8bce725e0 : SUCCESS in 7m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/822af3cced4546ada2b73544780b3717 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f6bb68cb8e148208eea2793f56a98e9 : SUCCESS in 55m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2b1969360314b4190f6e116591eabc0 : SUCCESS in 1h 27m 55s","accounts_in_message":[],"_revision_number":1},{"id":"19654f0f060a8b04e29e43a5ac643874914ffaca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-05 13:55:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"039fbb4df11efe220a8540c5b37a3ca6e8538843","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:36:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/80fc87c9c8bc4493a184029a3917a666 : FAILURE in 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a99479798debf404828289586a0ab718ee1e6aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 18:56:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eca49fa8a82465ebe510d5735ee48b3 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42054b58486f47099301b6fb3ee56d9b : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b81e7664b013445eab826d0792e0ce48 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74abdd47f0484af39187e230a01f371d : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8f93795ba064a89a7c4f09273cf43f8 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f788a608735419588e29f7f64dad618 : SUCCESS in 7m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8a0980cbc714fc8a58e0ad7db010b36 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93528880093340909661d142e55582a8 : SUCCESS in 11m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a4d310b7fceb436fb9afbeb152461720 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7fae0ae8682949c8aaab7a56bd824c3c : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f9187583ae4426381b8931553ea9c21 : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/656179e013b44f09a4eb92436b9ffe6a : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c95848167c64560bb21c291b979c06f : SUCCESS in 9m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1fc6611409724bc6a0caec2bef684975 : SUCCESS in 14m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a9c36e4349b1430596ce3394ee7ca96f : SUCCESS in 16m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/314a65a4f89d4ab3a6283bbed65bb2bc : SUCCESS in 5m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e4b4bb88321a4a8693c08c566f5a0565 : SUCCESS in 28m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7aaf2a82c275482da46f0614ae6f3bc3 : SUCCESS in 15m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1286778f329c4d6391d2e8a944d714e0 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/832a2394cdb047abacbde818e5f1b48b : SUCCESS in 19m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b335bb49a766474a87d8767e394e393b : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e07dea12a214fe0add3353fac63debf : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba5e734948c3430c815447421eaf832f : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27a9f55ee9b341b1abb2cf5e221229e1 : SUCCESS in 56m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e252fd2d86c4d0892db20535eb4acb8 : SUCCESS in 1h 38m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1301dc5a4305b83b32043d7ceaa8a4dce8b108c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-05 20:22:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92d1919acadee2f40b8ce287aff13f195ed4522b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:24:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c431a844aaa87711377a9af596fd9cf60a996ea3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 14:32:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00333b4d7c1c475bb1eceb1de393798f : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0df517441a4772a42301c02aa670fb : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/843b0616fbf2455b921f7b019563a401 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aec72748f3a423eb96790507f7d081c : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/051c1523426e4414ad2f28435185c2d7 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/424649a90755422ab7150ac7289f727d : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/172605bd12be41d19591318b7f56d2c8 : SUCCESS in 8m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a2d89a12eb5421795468e0e53f25072 : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ad02a6ce840d4bca86fc85d0af9bc9ef : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3a3b0352198540d68867280dc06c1ee5 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7a5d7c092c59408586b11fb9d7534c40 : SUCCESS in 8m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98ba81d292914791910e853f74013653 : SUCCESS in 7m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ccab9e75b3849198918fec9a65f7ffe : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b09d30225da40d1a4c2da7264c0ce20 : SUCCESS in 14m 19s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/24786a9c2bb44b0488c295905e677387 : SUCCESS in 4m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb3e6b3bf9ed4cc28f798854bc951f29 : SUCCESS in 26m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3bfce58d5e374411bbdba33fa5056d04 : SUCCESS in 13m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6e0cb49f85b447da4f4c656bc5a10b4 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/826a732c454b40ca9ab819438395ea74 : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e71998b94070483a8a36784a1f3b253e : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c852dd0cb63948dbbe2823b51f220f6e : SUCCESS in 6m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/54023cd7b0e64dc6945141162366e263 : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/198908785a4d4fa49d8e319376e965e4 : SUCCESS in 56m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e0170b4f9e54f7caca07c3108371377 : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":1},{"id":"69559c4516c578be19bf9cde372732bc001bf625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 14:32:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"32c354ef7e544f84f19817c8b05c6435083ed3cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 14:34:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2367cc094f4d16ae2140676ed4bac4 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"707a2bfb64cb3e19b76214e39f44aa8d2936304e","revisions":{"707a2bfb64cb3e19b76214e39f44aa8d2936304e":{"kind":"REWORK","_number":1,"created":"2021-03-05 10:07:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/778873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/778873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/778873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/778873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/778873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/778873/1"}}},"commit":{"parents":[{"commit":"8042d80030138de51aa011ce40146173e3550b2a","subject":"Merge \"Revert \"update constraint for ironic-lib to new release 4.6.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8042d80030138de51aa011ce40146173e3550b2a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-05 10:06:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-05 10:06:53.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.3.0","message":"update constraint for tosca-parser to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie1dd06525b161219de0d85ac8dad7aa0fcc8573c\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia4900fb8b37d3711c080e5d0c22b963495e2f852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/707a2bfb64cb3e19b76214e39f44aa8d2936304e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/707a2bfb64cb3e19b76214e39f44aa8d2936304e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
