)]}'
{"id":"openstack%2Frequirements~778875","triplet_id":"openstack%2Frequirements~master~I423f6ef4a2347bf3d0d4930e1a60df60f823d107","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I423f6ef4a2347bf3d0d4930e1a60df60f823d107","subject":"update constraint for monasca-common to new release 3.3.0","status":"MERGED","created":"2021-03-05 10:12:33.000000000","updated":"2021-03-05 16:52:17.000000000","submitted":"2021-03-05 16:50:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778875-new-release","meta_rev_id":"df5fb233a6d48f685021eea9ef576a835c3a3aa6","_number":778875,"virtual_id_number":778875,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 16:50:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-05 13:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-03-05 10:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-05 13:55:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 10:14:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-05 10:51:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-05 11:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 13:55:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f1b6fb160602b5e17cf3d9532eab01117b8243be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-05 10:12:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b87a1400af4c603b4a752ce649326db3580fb92","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-05 10:14:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fedc8de544548c3b20466b3b0b30cc0a46920f1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:51:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4ee74bef7ae98e11ddb34b835d372ad06c2aacf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5373dad2a707465c89e9caf40742dc9e : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35793ff85a9849b9ba075d98daaf5fd9 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc1d6c198a34383beacee8f11799684 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d693523524f4425f9d932ee459a78ce4 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cb49b663f8f47829c8f81c48bcb0f24 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9485829c45b84531a2a9b3c414bcd847 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8585e3dd9794f1cbbf84585554e074c : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d948ca97e92417bab1f151364c07275 : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42d98cdb4d954bf48c20c959dffa410a : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/06738a80e18142768182f8d7819ca6b9 : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6452738c95f44fa7b2101e004f3f0a43 : SUCCESS in 6m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9de812edc8c74abe869b5fb09db3c2ce : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed31071d70db4b619d7b4ef5b47a84f5 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49d780ff80ba400b84babc768a125e4e : SUCCESS in 16m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7238caae7910499cac2aeda0e3ab38dd : SUCCESS in 13m 57s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5ae7f7eb10ea4f79873ee95f877ef869 : SUCCESS in 5m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0e98b4cbd6084224a64dba81db4ad1c0 : SUCCESS in 29m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cdf9737b87514822bc4b60eff2c1c635 : SUCCESS in 17m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c374e63c60bb4027917593f4e7072680 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb30e3e1bd1b456dbdc3b2fe2e9f484e : SUCCESS in 20m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/25d46a1503f54ec6855a1fe59bd20c7c : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87ab066a13414c2da9bb2d066e1940dd : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/64f775bd4007472a9bedf53d0b5ea9d5 : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9e3fd405111496a9d5b0c6ad28a6d47 : SUCCESS in 59m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec1e4f94ac0244edb3f311c72a405c12 : SUCCESS in 1h 27m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f456bf71caea8f7a08d4cccc2c5a9c5ce56efcba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-05 13:55:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2209ab03695d05172f1c9ac3ae308a4f59b3bec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9a9b5508b0c82d687c6d2dde6f16d6861843cb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:50:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f0140005dd4e5d994488d3c32f0aad : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28301a605c241ceb5b48381a1e4ed89 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a3f59fd4f1e4893949a31ece78df574 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/990d6a05953f44b7b2aa7573aa553268 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4811bf8a379479ba1cd03f7948b3640 : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8deed84c7054ae095bd4385330dc79f : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a391c57bd8a4ae8949ba2cfbdcf6400 : SUCCESS in 10m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa3bd46cc7fe4df7b2259f27b22867d5 : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d7cfab920c94562a4e9d6e6090152df : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/309360299b8d41d5a1539f8fbb321e60 : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/415f0fd2ba8e4f24b39d28fb6053f6e0 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15ac93418ddd466c88c0322a9a2b46e9 : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/732df666a1394d95bb6fe2d0d3284c9a : SUCCESS in 19m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cf94ed19ba014877826a0dc2ed895872 : SUCCESS in 18m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8487b6d868e24abbab58b9ad62b6c3fd : SUCCESS in 5m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e9ae7de50dd448d9b0fe1b62c3263b51 : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a2d00f8443e44f138c6d309ab610540d : SUCCESS in 13m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a79fd18344b44df1b73e1e4cf8784ba4 : SUCCESS in 6m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80cf680babdf4b979252622fd7e0d199 : SUCCESS in 19m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c44703b18cd347bba8def723b43c0130 : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a993fcb50c54b9a8293cd6396aee507 : SUCCESS in 7m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d381f40c08de491aa2b558a2e08e6b24 : SUCCESS in 9m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24b7195d3b2946999fe21e77045ac55b : SUCCESS in 48m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95ecfd68be184af1b640321043219731 : SUCCESS in 1h 50m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b5aedc9c5b23efa978703beda64703b2b741236f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:50:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df5fb233a6d48f685021eea9ef576a835c3a3aa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:52:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8fa334517f411c8b6583ed88ccec96 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a515e9b69529f4841be5a7600bdcd4c681ebafb9","revisions":{"a515e9b69529f4841be5a7600bdcd4c681ebafb9":{"kind":"REWORK","_number":1,"created":"2021-03-05 10:12:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/778875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/778875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/778875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/778875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/778875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/778875/1"}}},"commit":{"parents":[{"commit":"8042d80030138de51aa011ce40146173e3550b2a","subject":"Merge \"Revert \"update constraint for ironic-lib to new release 4.6.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8042d80030138de51aa011ce40146173e3550b2a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-05 10:12:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-05 10:12:22.000000000","tz":0},"subject":"update constraint for monasca-common to new release 3.3.0","message":"update constraint for monasca-common to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Icfb8c0800109872ac92d574aecabb3cb1da3a195\nmeta: release:Code-Review+1: Martin Chacon Piza \u003cmartin@chaconpiza.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I423f6ef4a2347bf3d0d4930e1a60df60f823d107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a515e9b69529f4841be5a7600bdcd4c681ebafb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a515e9b69529f4841be5a7600bdcd4c681ebafb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
