)]}'
{"id":"openstack%2Frequirements~779135","triplet_id":"openstack%2Frequirements~master~Iaee5cfca77ce83514c3359301126592d45460a4b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Iaee5cfca77ce83514c3359301126592d45460a4b","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-03-08 06:29:42.000000000","updated":"2021-03-08 09:27:29.000000000","submitted":"2021-03-08 09:24:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779135-openstack/requirements/constraints/noclob","meta_rev_id":"da3e8aed94cb84a0758d118344553751ad1aa17a","_number":779135,"virtual_id_number":779135,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 09:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-03-08 06:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-03-08 06:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 06:32:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-03-08 07:08:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-08 07:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe4f3abfc1425068cdeffff41f104edff15c0cc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-03-08 06:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6e0710f3bb28d2ff9fad872434f1c7cc4b1c8d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 06:32:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6221a38ac79f19edd16f12b367dab4c10bf6cdc3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 06:38:44.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"8b8a2b02d65891637ca824bc5ae3405f450d056e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-08 06:38:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"646862d669b449f5933e16da6f2aea0e7a4e0bec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:08:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/58d9fb0335194667822876f289595a7b : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f0e5c6dee332c6bb8134f1616a9d176be6617ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:38:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496c6515e9ab40d8a267621c039af33b : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b164d27820924097a0e34fd39ed9f9a7 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f20104a3c2194ff0b5ecdfe480f9ae28 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/247cda721d2a497e96f0c50e11ef9767 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/734376d03fb64e27a109b62f307c1818 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f61668d4e82149d39c009ffe79f61a9d : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45b6fb6351c44f0bb1ff38b121a35474 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4faca45e261454ab2f6bffb113aeb91 : SUCCESS in 9m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/51e78d095c2c428bac3919da3b961497 : SUCCESS in 12m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2baf7cf28c6d48f18bf40411e07b9fbe : SUCCESS in 11m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/20d325cbcff043ee90f9b77e7128989d : SUCCESS in 8m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa2e170d73884ff0ac3545842fb54e7b : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6243cd67849458ea4705a9662725a7f : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eab8fde4b3424b9d970d22d5347c1767 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8385b104502347fca67ebdf2912deb5b : SUCCESS in 14m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/678ab5eaba99421981b27d1d9a78a561 : SUCCESS in 7m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/761b8d8c138c48d1a11c3b53ec5e9a0a : SUCCESS in 34m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ca2377fde074a35b3deb3b6820f60ba : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60736ce749f54425a74e5e01afa54280 : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f56df36573bf4531a2d81f27eece4ad9 : SUCCESS in 16m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/554a8674c2484df19632eaaaa8bdeb83 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5138344a2d3d448eba1c43b2991f7a64 : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1416237c36f49708f4524243abef1ea : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90739f2586254223888910690e092fed : SUCCESS in 55m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70d53b16686c4c56ad99132bbd46ecf0 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":2},{"id":"3471ff84642c8a881b7f4cb45043d303ab888ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:38:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3d8abcbd41edb3c954d0697332ec9d6dff82f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:24:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6408a902084646baa0ea69f2446417bd : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9143ec402047a7b3a48ddb25d2e9bd : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab271a0809ed41a7aa54c519c3126375 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd19662f0524b2eb1c95dd8830fc5bc : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc72eb26c7924a8ebedd0b2876585ca5 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7ae1ad758cc4928ad726edf0d509c77 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5cb8eb8efbe1450e86ff8260e169693a : SUCCESS in 9m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38732e40f03344f099918fa009df9ee7 : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/468b100df6414fafa0249fdf18255255 : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/954f6226286d4603906ea7de48d09e46 : SUCCESS in 6m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88e44c5be1584912955870942745fbf8 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7ed0257f74b445195f4f7915ec5b437 : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7812b04bb02849d396036dd47fde58ef : SUCCESS in 9m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2254890e08fd4e9cbf69b16aa274cb0b : SUCCESS in 14m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9dfde008eb4b4a8f9edaa846a7a023d8 : SUCCESS in 5m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4004be6d7d7148ecaee98d2fbe3fe01f : SUCCESS in 32m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3d15a3379507498bb81dc31e33953c1a : SUCCESS in 22m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad9f5e6a38cc4d3b9d7921d5ff0bf53d : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55f81a11998745818558e58a0e9f416c : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8b3c469748fb4376854f15830da2aba7 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e917063bcf1a43628ab32fb9b0d238e5 : SUCCESS in 6m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9a73a2f1629e4012b24e6c8af92f8b16 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9fe270c3d9a41f3bbe058eab00b750c : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e549ff5657a54e5b97aaeb956f9547c1 : SUCCESS in 1h 43m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f43deaa1ead0e27d37ef6f3716aa75247a9515c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:24:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"da3e8aed94cb84a0758d118344553751ad1aa17a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:27:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c81da1dbc84de58c73f84f74382ab2 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79ff19fa57044575ea93b57c29c0e0b46028989d","revisions":{"5b2a5e57f8b854eef298ba19e8384e1585dd1783":{"kind":"REWORK","_number":1,"created":"2021-03-08 06:29:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/779135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/779135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/779135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/779135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/779135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/779135/1"}}},"commit":{"parents":[{"commit":"ac8bcb00122ccd4c4c8cc22612c14c3c37624026","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac8bcb00122ccd4c4c8cc22612c14c3c37624026"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 06:29:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 06:29:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaee5cfca77ce83514c3359301126592d45460a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2a5e57f8b854eef298ba19e8384e1585dd1783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b2a5e57f8b854eef298ba19e8384e1585dd1783"}]},"branch":"refs/heads/master"},"79ff19fa57044575ea93b57c29c0e0b46028989d":{"kind":"REWORK","_number":2,"created":"2021-03-08 06:32:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/35/779135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/779135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/779135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/779135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/779135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/779135/2"}}},"commit":{"parents":[{"commit":"ac8bcb00122ccd4c4c8cc22612c14c3c37624026","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac8bcb00122ccd4c4c8cc22612c14c3c37624026"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-03-08 06:29:41.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-03-08 06:32:38.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iaee5cfca77ce83514c3359301126592d45460a4b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ff19fa57044575ea93b57c29c0e0b46028989d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ff19fa57044575ea93b57c29c0e0b46028989d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
